)]}'
{"id":"openstack%2Fglance~747470","triplet_id":"openstack%2Fglance~master~Ic933f170d43b290fd819e8527ecb60be7f7f3f89","project":"openstack/glance","branch":"master","topic":"task-heartbeat","hashtags":[],"change_id":"Ic933f170d43b290fd819e8527ecb60be7f7f3f89","subject":"Make test_copy_image_revert_lifecycle handle 409 on import retry","status":"MERGED","created":"2020-08-21 19:53:47.000000000","updated":"2020-08-25 15:42:24.000000000","submitted":"2020-08-25 15:39:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747470-1598369954782-8296442f","meta_rev_id":"8f1f5d77ede2bad63b01f5d0b5dade2cd1cb8522","_number":747470,"virtual_id_number":747470,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-24 13:41:53.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-25 15:39:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-25 13:55:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-08-24 16:21:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-25 13:55:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-08-24 16:21:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-24 13:41:53.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-24 16:21:18.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-25 13:55:57.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-08-25 15:39:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d8494f650fe32ea2ad47becb6c3b89bacc283fec","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-21 19:53:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6109f156da6d61b1f8ab410ba72208fd4fb51849","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-21 19:58:48.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"7d68a58e6ea791e08879316f20104807b1a22821","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 22:14:58.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/048e650a1a334a5faf92d2054a45a234 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/742b5a8071d8418d8fb2cdb21a1b7483 : FAILURE in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50cfd9c7dddd41f4bc15c80b8444aecd : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80b130bbb34d4edca9c3992a61742401 : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf5ea699b91e4eee95862a91247ad95f : SUCCESS in 11m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cbbd2886cd8e4c8d9ee179c5c74a3ac7 : SUCCESS in 10m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a3ba50ed7bf7408d92a66440af9fd3b0 : SUCCESS in 12m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/63bb6b2a7bb347d8bff07abe592da3a2 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":1},{"id":"010edd87c78e64fed0abefd4b0af8d6cfb343c89","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-24 13:41:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc60d14ca91f83d37ca3d5a7a37a9d0d0302448e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-24 13:41:53.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"130596d4e038f9f97baf2d867ba078a115341a04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 15:40:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4695557b5b043268fe33b6b1048e853 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed4be3ee458d45478c92fa89f8ab620a : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2e5a1bf718b4cd8a2ec35641ec4d6ca : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e241296b64742aba938131205e98000 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fe5a06f97da40ebab9c99e7e1e9a33e : SUCCESS in 6m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3c6c844305d4673be6b385d15fe6b87 : SUCCESS in 7m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/089081153b004f55a628d64fd975e3e6 : SUCCESS in 7m 45s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b4ef49c79b2640dea8208ba07655cdf3 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e3e62b9b663e27b1e43cfedbc01edc863dbaefc6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-24 16:21:18.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\nLooks good approach to me to wait for lock to get busted before next import call.","accounts_in_message":[],"_revision_number":2},{"id":"f42a6990fc495bc1356b7e4ba8e02e30f6d57f05","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-08-25 13:55:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nMakes sense to me.","accounts_in_message":[],"_revision_number":2},{"id":"84cc71e4a7c5ebd2c56c333772ece0f3d348815d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 13:56:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1c3b39337dc7b32a9a29c0062a98164007ca1f8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 15:39:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b59f53f02584c528445f6364a595927 : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d74150677004446eb3c5d04573d81ec2 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/037948183afc4a3a93e461858e820ab0 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de9933601c63484bb42ba39627eee0fd : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e85dae0e34eb41d2a9c7dbe3d1c942e1 : SUCCESS in 8m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ba7ba1fc74cf48029291bd0b14c27978 : SUCCESS in 9m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1682441e4c854a70b336f982a2f0678d : SUCCESS in 12m 04s","accounts_in_message":[],"_revision_number":2},{"id":"97a92a1ea88dab7e52c7c614e48d99ac5af995fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 15:39:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6b4a1b4846886bfeca78a1a842c213e3c49a77ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 15:42:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8771b3f9cb114b3c9ed8dea287655022 : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8c7342cbc15fdbcbafa1ead9aa183e45862cf177","revisions":{"588c5ae5e4012753089b4694837d9473352ad738":{"kind":"REWORK","_number":1,"created":"2020-08-21 19:53:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/70/747470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/747470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/747470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/747470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/747470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/747470/1"}}},"commit":{"parents":[{"commit":"1580b4d82105381541ffdb1f1609383c2a2da636","subject":"Merge \"Poll for final state on test_copy_image_revert_lifecycle()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1580b4d82105381541ffdb1f1609383c2a2da636"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-21 19:49:37.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-21 19:52:05.000000000","tz":-420},"subject":"Make test_copy_image_revert_lifecycle handle 409 on import retry","message":"Make test_copy_image_revert_lifecycle handle 409 on import retry\n\nThis test is very racy in general, and specifically on the post-revert\nretry to do the import, it may start the next import before the previous\none has finished reverting. This patch makes it retry a few times with\ndelay if an HTTP 409 is received so it is tolerant of that situation once\nit becomes possible in the subsequent patch to add import locking.\n\nChange-Id: Ic933f170d43b290fd819e8527ecb60be7f7f3f89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/588c5ae5e4012753089b4694837d9473352ad738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/588c5ae5e4012753089b4694837d9473352ad738"}]},"branch":"refs/heads/master"},"8c7342cbc15fdbcbafa1ead9aa183e45862cf177":{"kind":"REWORK","_number":2,"created":"2020-08-24 13:41:26.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/70/747470/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/747470/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/747470/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/747470/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/747470/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/747470/2"}}},"commit":{"parents":[{"commit":"1580b4d82105381541ffdb1f1609383c2a2da636","subject":"Merge \"Poll for final state on test_copy_image_revert_lifecycle()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1580b4d82105381541ffdb1f1609383c2a2da636"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-21 19:49:37.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-24 13:41:13.000000000","tz":-420},"subject":"Make test_copy_image_revert_lifecycle handle 409 on import retry","message":"Make test_copy_image_revert_lifecycle handle 409 on import retry\n\nThis test is very racy in general, and specifically on the post-revert\nretry to do the import, it may start the next import before the previous\none has finished reverting. This patch makes it retry a few times with\ndelay if an HTTP 409 is received so it is tolerant of that situation once\nit becomes possible in the subsequent patch to add import locking.\n\nChange-Id: Ic933f170d43b290fd819e8527ecb60be7f7f3f89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c7342cbc15fdbcbafa1ead9aa183e45862cf177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c7342cbc15fdbcbafa1ead9aa183e45862cf177"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
