)]}'
{"id":"openstack%2Fglance~747991","triplet_id":"openstack%2Fglance~master~Ia78d101ae3702d9f761ca8f04b4b10af4a6d84fb","project":"openstack/glance","branch":"master","topic":"bug/1891190","hashtags":[],"change_id":"Ia78d101ae3702d9f761ca8f04b4b10af4a6d84fb","subject":"Disable wait_for_fork() kill aggression if expect_exit\u003dTrue","status":"MERGED","created":"2020-08-25 17:32:03.000000000","updated":"2020-08-28 03:36:24.000000000","submitted":"2020-08-28 03:34:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747991-1598585681064-46488824","meta_rev_id":"162ed6a0cc6a325a2bda903504feb72590b80363","_number":747991,"virtual_id_number":747991,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-25 17:37:41.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-28 03:34:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-27 13:53:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-26 06:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-27 13:53:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-26 06:59:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 17:37:41.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-26 06:59:39.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-27 13:53:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-28 03:34:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f17b83e73e802534c3b3c678d1224cea1b86c50","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-25 17:32:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ddc68159c9844cc7c6cdef8a8f16aab35bdf4d6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-25 17:37:41.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"f4b193de4effda8aa0f809742b83346274e765b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 18:42:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2618e46a9596472e837ca60fb2cb43b4 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d2ef173296f4c5f9f89505e7e88b296 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc0817c79a814e589fbeaa4bae5348d3 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37c0e60ef0ac4fb39f8ae73094a4ea6b : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b72adfa76ee44964ad265f8b16fb8521 : SUCCESS in 7m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4bd7cef55c7542fa94668db5b8603e73 : SUCCESS in 9m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/141547675845462dad40616838108b1d : SUCCESS in 8m 58s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5425b8bd38104f26ab7d353b714f2dbd : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b3b492a6e97793cf75d435c4afa602a8f9baaaa2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-26 06:59:39.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"9967555872c25145d2713e432e7348b808169ae8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-27 13:53:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nEasy enough.","accounts_in_message":[],"_revision_number":1},{"id":"78c247fb4db34486662d8f147616e5814df4906a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 19:38:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f585f58bc29f4ec8c64962e80272570d72a422d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 03:34:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29a59d42e45c43d1bf53bbf9f899efa4 : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b55bee91bad64e27893ea7cc4ed96967 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e9cc8a54bb04183a73a904fc89a7d85 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86847bdec6b647c897b6ec2a2feba56e : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4be587488bd34ce19457effc40d1c198 : SUCCESS in 8m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1f82eaeb112545f2bb5dd22c80a23275 : SUCCESS in 14m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ac2d2a6769214c93b8364f60804160ed : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"fe34a0723f8016d6291802f2f9db1eeba8e67206","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 03:34:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"44acc6638596938bc8cc03e61620c4c4b36b9c18","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 03:36:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2683730e85684f6fafe979fdd257976b : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76667323d7203a87447269e4f834f7d9339244fd","revisions":{"76667323d7203a87447269e4f834f7d9339244fd":{"kind":"REWORK","_number":1,"created":"2020-08-25 17:32:03.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/91/747991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/747991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/747991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/747991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/747991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/747991/1"}}},"commit":{"parents":[{"commit":"3f6e349d0853a9746d0d744bc3eb0b2baa1ddff9","subject":"Implement time-limited import locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f6e349d0853a9746d0d744bc3eb0b2baa1ddff9"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-25 17:29:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-25 17:31:45.000000000","tz":-420},"subject":"Disable wait_for_fork() kill aggression if expect_exit\u003dTrue","message":"Disable wait_for_fork() kill aggression if expect_exit\u003dTrue\n\nIn some cases, we spawn a process that we expect to run to completion\nand exit. In those cases, disable the force-kill logic in wait_for_fork()\nto avoid killing those processes too early.\n\nRelated-Bug: #1891190\nChange-Id: Ia78d101ae3702d9f761ca8f04b4b10af4a6d84fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76667323d7203a87447269e4f834f7d9339244fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76667323d7203a87447269e4f834f7d9339244fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
