)]}'
{"id":"openstack%2Fglance~748002","triplet_id":"openstack%2Fglance~stable%2Fussuri~Iac75956e314ec6f41db18430486bd8be9754e780","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1884596","hashtags":[],"change_id":"Iac75956e314ec6f41db18430486bd8be9754e780","subject":"Make import task capable of running as admin on behalf of user","status":"MERGED","created":"2020-08-25 18:57:25.000000000","updated":"2020-08-26 23:13:15.000000000","submitted":"2020-08-26 23:11:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748002-1598483504670-4284198b","meta_rev_id":"baa738b28ba7cc517bced509290f7fd101cef392","_number":748002,"virtual_id_number":748002,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 23:11:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-26 20:53:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-26 06:15:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-26 20:53:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-26 06:15:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 06:15:28.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-26 20:53:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-26 23:11:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"073ace53cfe4be5c5de797e751668e4903a741ed","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-25 18:57:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1364db86abad0991ab58ccaab166fc5058a0e15c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 20:08:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d1dac539f5eb4e0b9b7ee8922007077d : SUCCESS in 1h 00m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/52e8f16332004ea39d7c2010c3e7f07e : SUCCESS in 1h 09m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cee949170f44210b00b05ba3f23ded1 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/518ace3cff8545108d8d548ff89cb1cf : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d72dc00382f4684ae72715c0871d9e8 : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0590a89fc31f4bd696fbafcfe1147ec0 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0200f7b36674434bad6cf1dab7ca535 : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/48845e9c1fe34fc9b48d3b4d65c377f3 : SUCCESS in 11m 30s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/88b370cdbf0f4710affaf0fbd6bf2c06 : SUCCESS in 7m 36s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c25773a7a0ec4a8cb870ca39f9c8d437 : SUCCESS in 5m 37s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3a26ef6c333144d7aa6258e9229e606e : SUCCESS in 1h 01m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89304f8e86424bbda2854bb225a970c6 : SUCCESS in 59m 20s","accounts_in_message":[],"_revision_number":1},{"id":"74b4eb46c8fe1f96a517b38fcdb469a51b427c7a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-26 06:15:28.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"7d774250628ba655a0050079b27d56a6ac619a6c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-26 20:53:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nModifications look good.","accounts_in_message":[],"_revision_number":1},{"id":"7f0894f2f7fb682581fefb31e5a80865fd3ee4bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 20:53:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"799fa32d65145aa5dfbc31feaa227fc530cc5e3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:11:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a51e76c129e9bda98b2aa7b86a83a3ed29d549a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:11:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/86327fe7da8b47c78b7a3b51d114f745 : SUCCESS in 52m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/be169d0b2c164628bcb80f10c56a4d2e : SUCCESS in 1h 15m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff9277b90568403db60d0f4a0d0fee93 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b2152e3b39e4fffab3e50e517935fa7 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3c8fbc5898a4af7a4aad715d086f380 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e61aef8d96734d1ea48d731564c9dccb : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eada1fc4e7e640c880ed13db0b0a523d : SUCCESS in 9m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f1413801986f40f89903e26124435050 : SUCCESS in 8m 37s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/fb6eb4a4fdbe47aabd2152cd8616b848 : SUCCESS in 9m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c455dee488d41e5b04d793fc6aa4a11 : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":1},{"id":"21939492ea19f5ac2b791bffc23671d61ecb2986","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:13:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d6ef3195ddc4a149c70147940db8a41 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb08fbae0f0883fd64babc1215aeb6ba46c0c15b","revisions":{"eb08fbae0f0883fd64babc1215aeb6ba46c0c15b":{"kind":"REWORK","_number":1,"created":"2020-08-25 18:57:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/748002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/02/748002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/748002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/748002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/748002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/02/748002/1"}}},"commit":{"parents":[{"commit":"168fdb7f304bf16ce0170f517442504617a29fc6","subject":"Add a functional test for non-owned image copying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/168fdb7f304bf16ce0170f517442504617a29fc6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 18:51:37.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-25 18:26:41.000000000","tz":-420},"subject":"Make import task capable of running as admin on behalf of user","message":"Make import task capable of running as admin on behalf of user\n\nThis makes the api_image_import task capable of running as an admin on\nbehalf of a user if so authorized by the API. It includes a new object\ncalled ImportActionWrapper which provides a bundle of utility methods\nwhich can be run either against the user-authorized or admin-authorized\nImageRepo passed in from the API. It encapsulates all the actions\nwe are able and willing to run as an admin for the user.\n\nThis is currently not drivable by the API because the policy check is\nstill statically defined as \"admin or owner\" but this change is\noffered without any needed modification to the functional tests to\nprove that it does not regress existing functionality. The following\npatch will introduce a more robust knob for allowing users to do this,\nand it brings the functional test changes with it.\n\nChanges:\n - Removed the change to the images API to pass in an admin repo\n - Reinstated the ImportActionWrapper usage that was removed from\n   backport commit 2b3ce07c32624f9987df87520c5ff8d538c8e5fd\n\nChange-Id: Iac75956e314ec6f41db18430486bd8be9754e780\n(cherry picked from commit 2fd0c2573311db6a50228ed22adf8bc2b5638cfe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb08fbae0f0883fd64babc1215aeb6ba46c0c15b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb08fbae0f0883fd64babc1215aeb6ba46c0c15b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
