)]}'
{"id":"openstack%2Fglance~748003","triplet_id":"openstack%2Fglance~stable%2Fussuri~Ic933f170d43b290fd819e8527ecb60be7f7f3f89","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1884596","hashtags":[],"change_id":"Ic933f170d43b290fd819e8527ecb60be7f7f3f89","subject":"Make test_copy_image_revert_lifecycle handle 409 on import retry","status":"MERGED","created":"2020-08-25 18:57:25.000000000","updated":"2020-08-27 02:18:30.000000000","submitted":"2020-08-27 02:16:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748003-1598494603991-fda0f000","meta_rev_id":"878b4e889bc13799874f1d3d6cd81b57f66afcce","_number":748003,"virtual_id_number":748003,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-27 02:16:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-26 20:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-26 06:23:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2020-08-27 01:53:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-26 20:53:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-26 06:23:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 06:23:39.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-26 20:53:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-27 01:53:13.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-08-27 02:16:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"436427415e6e7bcf9e4cfac44ea43bbfd6b6c38f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-25 18:57:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e5c48c3015b8a9407e26e67dc3ded588e9b11a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 19:16:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8778fa96e5c7454c8960f0fc37fa5a7c : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5284301c3c9d4656849654e0511b3a30 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f779fb9222f24489b68d0edca53c6ea0 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d786dc01a53a40e7919d6fd15a85e9d5 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f157136338404d3496d0e9919399998b : SUCCESS in 11m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e91cde95335e4b9886538d812dddb551 : SUCCESS in 11m 10s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/fb9e72ef72ca40929e950cf0a546dc35 : SUCCESS in 7m 41s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/883a227bc3e341c6aaef8c0cc0973d87 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"13d8d24a03e08b9391677f711a6acfd76a821cb1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-26 06:23:39.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"81c392a2d2de954965254612d69c45b360fed661","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-26 20:53:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22290cc4e6ca2706dc64f534f31d57aeda9db22e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 20:53:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b627caa3b048ad4aff8c2b9a7d9c72222058b8e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:11:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8d0bac4a91448b6bb77e7c3b1c1c024 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/507afa37c4e942b6863d84cf3ca0993e : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dac0b4811287478f9eff24f30237afb1 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f4f1656f30042cea71150b3af70ebff : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77bd11b53bfe44c19f20afd4fee7bf01 : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb0e02c4178248c982c559f86399e9e2 : TIMED_OUT in 31m 07s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4daeb4c110a74c65a06e1d92fd57dfb0 : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f536a03406acd75ebc52ffab4585001cf908b728","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-08-27 01:53:13.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e85e14403b9445313e1af191f3d88dd1af102814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 02:06:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/708da305fb49495eb2256fdcad33b6df : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a436252efce41ebb4cdc03d370bd664 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4230c65c982428e9bf4d7ff76a35722 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30ca4225d24d4bff8bd50a4525b0d136 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3fcfa05ca6e4d7b9f54dabfabe9dabf : SUCCESS in 5m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6d68da63c89040dea2031a09817055de : SUCCESS in 7m 47s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/71150d5066064cdc9c64594fd2796f25 : SUCCESS in 12m 31s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/270c9bdd45e4401b80d28b3b20680a6d : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"04dcf1b7eb34ff7b92f59031a7f508c2f2f89a4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 02:07:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a3567e55b7275db8ced8083ac28639efd5141d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 02:16:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df16841220c84f7484cfe908082ed1ec : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ed88b774af249318009d3e56e759c31 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecbc1715fea743d7926b848d3cbcfdd5 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3d2698789784869b010a98db55f794d : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f839a964dd5b4d16940aac5436a46ea7 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f3d535f0a44c4908a827991de457d8f6 : SUCCESS in 8m 23s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/49d7784c311d4d98a9a3b6ee59644435 : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":1},{"id":"56dff3ab79d1d4195ae8ecbfbe2ad71ce23f7732","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 02:16:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"54e8b9a3f36688d9436e9d363b72c69ff5bbc09c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 02:18:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980d8e6ab51c4bf7a1534b72fa47625d : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0895d19197ab6313ef80294fe33807a4a71a44ea","revisions":{"0895d19197ab6313ef80294fe33807a4a71a44ea":{"kind":"REWORK","_number":1,"created":"2020-08-25 18:57:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/03/748003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/748003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/748003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/748003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/748003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/748003/1"}}},"commit":{"parents":[{"commit":"eb08fbae0f0883fd64babc1215aeb6ba46c0c15b","subject":"Make import task capable of running as admin on behalf of user","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb08fbae0f0883fd64babc1215aeb6ba46c0c15b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-21 19:49:37.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-25 18:27:36.000000000","tz":-420},"subject":"Make test_copy_image_revert_lifecycle handle 409 on import retry","message":"Make test_copy_image_revert_lifecycle handle 409 on import retry\n\nThis test is very racy in general, and specifically on the post-revert\nretry to do the import, it may start the next import before the previous\none has finished reverting. This patch makes it retry a few times with\ndelay if an HTTP 409 is received so it is tolerant of that situation once\nit becomes possible in the subsequent patch to add import locking.\n\nChange-Id: Ic933f170d43b290fd819e8527ecb60be7f7f3f89\n(cherry picked from commit 8c7342cbc15fdbcbafa1ead9aa183e45862cf177)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0895d19197ab6313ef80294fe33807a4a71a44ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0895d19197ab6313ef80294fe33807a4a71a44ea"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
