)]}'
{"id":"openstack%2Fglance~748004","triplet_id":"openstack%2Fglance~stable%2Fussuri~I4af4202d6650b60aebbed813b7640cf018466e12","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1884596","hashtags":[],"change_id":"I4af4202d6650b60aebbed813b7640cf018466e12","subject":"Add testing for _CompleteTask in api_image_import","status":"MERGED","created":"2020-08-25 18:57:25.000000000","updated":"2020-08-27 02:19:37.000000000","submitted":"2020-08-27 02:16:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748004-1598494618855-1545f216","meta_rev_id":"5bce81be5655cd68ad161d0052eef6cdfe9dd21e","_number":748004,"virtual_id_number":748004,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-27 02:16:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-26 20:53:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-26 06:24:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2020-08-27 01:53:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-26 20:53:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-26 06:24:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 06:24:00.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-26 20:53:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-27 01:53:37.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-08-27 02:16:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c1bc3109e371a319b58cab1a71d99b3c883c5c1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-25 18:57:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f744a72bfc0d2affe46e8fe461504f11d7c5cb16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 19:13:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2621948bb5a4cfcaaba31e4361ef409 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb45745474d347318a256492bdba015a : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17362c02de4e4b03a9311773c4c29c0c : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6af1a5653dc34296b4f6ac450a87c92e : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50850ae0bd794ab6953acb0b62cd9a1d : SUCCESS in 7m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bbc6abbb765d41a196da13da2dfebdfa : SUCCESS in 9m 37s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/48e600be8cd54350b09f273ddf905bb1 : SUCCESS in 12m 50s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/03dbe8e4963f4aeb99024b14aee67c54 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8bd9ad6d58ee49a3cb37ecfd0aa2d54158ffd18c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-26 06:24:00.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"28351b0156fd51cbb1b58be370fdb2648e42ce4d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-26 20:53:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nGood test coverage.","accounts_in_message":[],"_revision_number":1},{"id":"77af78ca07cc810a81d34ba1025448fee990c17e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 20:53:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1381a7f61dd678ae3b9b7df2419c79786e2ea2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:11:46.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"ff82c12d52a42ad016d1b3dd7dc118044d1f63da","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-08-27 01:53:37.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"facd45649b41e979c51603612738c1e3a01aa456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 02:07:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/788febf9ad8b4417afca88fb20f55855 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc02d3c2d0c4d299645d93bd2e3378f : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16f968e583dc4f408f3e67bcb3ff81de : SUCCESS in 6m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a53650ae31f413fb2975a0f2ea7ecaf : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cae7b3f7cb34e5cb0e52a595ccbc089 : SUCCESS in 5m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/48b0d72af6d84f6bae0542b40b57f9ff : SUCCESS in 7m 27s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b525d05e35e24ef78c22115a6a4a5c66 : SUCCESS in 12m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2f1c7024ea1246cd949569ab8f205755 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ca6a6246ed17039f6bed66b7cec5073c4e497130","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 02:07:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15c370ff6d802c17a299ce9fc0fbe7ddb6db66c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 02:16:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15b3b5063e8a48d5b7ad2c9e20fef2eb : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0007c3006c1a48469aa6352e8ffbbc27 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f4d386b3756411aa29392a4e678638b : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eefdbaf843f7442282581423b3df674a : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3f3ebf6a8e340dabcf2c26d314821cf : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bccbee77064947b78de3e18abb97b143 : SUCCESS in 7m 21s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6a8400c817b042209ed6074614150e77 : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":1},{"id":"20b77ae855a1ef0c6aa2bfc9ea7c1c714169684c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 02:16:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c13d636102dee27d977abcd909952ac756084428","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 02:19:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c315eba6ae64c89a1b01439a2ec6a5b : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea34be42632b82b724adc9e6eee22d3367da6cae","revisions":{"ea34be42632b82b724adc9e6eee22d3367da6cae":{"kind":"REWORK","_number":1,"created":"2020-08-25 18:57:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/04/748004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/04/748004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/748004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/748004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/748004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/04/748004/1"}}},"commit":{"parents":[{"commit":"0895d19197ab6313ef80294fe33807a4a71a44ea","subject":"Make test_copy_image_revert_lifecycle handle 409 on import retry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0895d19197ab6313ef80294fe33807a4a71a44ea"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-28 15:32:46.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-25 18:33:26.000000000","tz":-420},"subject":"Add testing for _CompleteTask in api_image_import","message":"Add testing for _CompleteTask in api_image_import\n\nThis critical task has no testing. Add some before we add more\nfunctionality to it.\n\nChange-Id: I4af4202d6650b60aebbed813b7640cf018466e12\n(cherry picked from commit 2e11ecb15e63bac7b2de353ed77abf6a363d4158)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea34be42632b82b724adc9e6eee22d3367da6cae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea34be42632b82b724adc9e6eee22d3367da6cae"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
