)]}'
{"id":"openstack%2Fglance~748006","triplet_id":"openstack%2Fglance~stable%2Fussuri~I4b63706927b599eb343b614f0c396f699378e14b","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1884596","hashtags":[],"change_id":"I4b63706927b599eb343b614f0c396f699378e14b","subject":"Flesh out FakeImage for extra_properties","status":"MERGED","created":"2020-08-25 18:57:25.000000000","updated":"2020-08-27 05:59:13.000000000","submitted":"2020-08-27 05:58:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748006-1598507881860-436c11b4","meta_rev_id":"d4585a1f2f7910f0743762dca4bf50a1b42d79c0","_number":748006,"virtual_id_number":748006,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-27 05:58:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-26 20:54:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-26 06:24:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-26 20:54:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-26 06:24:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 06:24:46.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-26 20:54:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-27 05:58:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b01ba14874f3c9722760242db0b7116a917436d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-25 18:57:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"505e089948f75601d71f96931439ab97534c8106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 19:09:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ff82ed510b94efeacae7eac01afdb85 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6df2e33e0414f79884a229ca39b84a8 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a15b5c04bf94d63aa58e2bdeb405f7f : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d686bf56ae44433bbb5d7e5832408a3c : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65b9b4e087d24dea844a4220af33391c : SUCCESS in 8m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3c2a7c2d4d24afbbc6ea6f600c0c51f : SUCCESS in 8m 14s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4bcbf4c679fa4cd3bbbf46a0260d335c : SUCCESS in 9m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e4f7b51d07c34515b10dffa02ecf8777 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a119e27ab947eed8970d2689aa04672cb2524037","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-26 06:24:46.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"9f6da61e600a804515d16630b2893c3f2d341180","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-26 20:54:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nUseful change for later changes.","accounts_in_message":[],"_revision_number":1},{"id":"bd167f9c4352cd401944c38e113d929028640b46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 20:54:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5dc739d39ad1a08b4578e3c25ad698d418c75b87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:11:47.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"23d838b8ecc556daad0c52746b41ea62d69a5638","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-27 05:31:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"04c22afc6429fa54a8d31b2b58abebe1a7872093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 05:45:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91d7d22c174548ffb55fd4666ae81c79 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01bccdbcbb9d447dbf14a06c86a2e557 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e919f49e41bf421b9d31a420d5774f1f : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac42727f3a394f269f7a703fb5a3d7a0 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2526a084ac614cdd984aaab75e1061ea : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0fcc3fcab83146cea7d486b27bbe00e3 : SUCCESS in 12m 35s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e7e85ed1ef2440b89c628718ac1aea6e : SUCCESS in 7m 53s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6828788c37ae4eff95826bbf4ddf03ff : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8c3ee64cd7cf4d548197bbbf90e14c826c65d5be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 05:45:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e81a8dcb12d40e281df526567fbac43e1dc9907e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 05:58:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"21b1db434d8c2f359282a04dcb1727e75c0f6e76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 05:58:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f404956d7cab4fa0b28432c4d6bf05cc : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/256dc7b02f614b45ab30c50e7e18fe02 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/984a5721df7e4f65b4e5610ab7590648 : SUCCESS in 7m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2519ec7dcbb94aa0969ba9e9cd8b240e : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c820db2c529c4c6a8ab5f8b64de1c65b : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa03943bb49e454f9d241866dafb6428 : SUCCESS in 6m 53s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/893d86a48a76461f81fa27daf054ed48 : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ef20c87bb80c083a30bf4b3bee8768afcdde8aab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 05:59:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a2e5b6b98f84454ace463251845298d : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"663e027e50533909833b7ef11d82b9bdcda86499","revisions":{"663e027e50533909833b7ef11d82b9bdcda86499":{"kind":"REWORK","_number":1,"created":"2020-08-25 18:57:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/06/748006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/748006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/748006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/748006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/748006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/748006/1"}}},"commit":{"parents":[{"commit":"ec46cbca97308de672b2e91b0ae19bbe5617ed44","subject":"Add tests for _ImportToStore.execute()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ec46cbca97308de672b2e91b0ae19bbe5617ed44"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-29 15:13:36.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-25 18:33:45.000000000","tz":-420},"subject":"Flesh out FakeImage for extra_properties","message":"Flesh out FakeImage for extra_properties\n\nThe FakeImage in test_images_resource.py is missing a lot of standard\nproperties. That becomes a problem as soon as you add extra_properties,\nwhich gets delegated to for any missing property by ImageTarget. Thus,\nif that fixture has extra_properties defined on it, any other base\nproperty (i.e. checksum, size, etc) will be looked for inside that dict\nif it is not defined on the fixture itself.\n\nThus, this adds fake values for all those properties and adds\nextra_properties\u003d{} so that subsequent patches can put things into\nextra_properties without a bunch of work.\n\nChange-Id: I4b63706927b599eb343b614f0c396f699378e14b\n(cherry picked from commit 7c91a177edeb3782e99993d0568cddf31bff96a8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/663e027e50533909833b7ef11d82b9bdcda86499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/663e027e50533909833b7ef11d82b9bdcda86499"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
