)]}'
{"id":"openstack%2Fglance~748007","triplet_id":"openstack%2Fglance~stable%2Fussuri~Ifdb711cb241ef13eccaa5ae29a234f2fe4a52eb8","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1884596","hashtags":[],"change_id":"Ifdb711cb241ef13eccaa5ae29a234f2fe4a52eb8","subject":"Add image_set_property_atomic() helper","status":"MERGED","created":"2020-08-25 18:57:25.000000000","updated":"2020-08-27 08:09:26.000000000","submitted":"2020-08-27 08:07:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748007-1598515662431-4284198b","meta_rev_id":"d031b62c0c7a6431646281dee51ec167e403b7af","_number":748007,"virtual_id_number":748007,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-27 08:07:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-26 20:58:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-26 06:25:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-26 20:58:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-26 06:25:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 06:25:00.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-26 20:58:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-27 08:07:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52afbf8c3ac36efb4f4c82c3da70bd61b495a812","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-25 18:57:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42226bce30c3b95a0d7762af9d6781c794d454cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 20:16:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/96f15b53542b4461b338f26fe0a8b56d : SUCCESS in 1h 12m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d46992551f5b44eb852360886b7f6e5c : SUCCESS in 1h 07m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee1e53b95bfa432190bfa071e17007b2 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ed92d3b5ef04f4297346eb94cc3b399 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d28b149727d49c59cb98abb53efb195 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d45bf4154d84493b346931c26269fe8 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f5b47aa9cb7461287287bb2aa245ed9 : SUCCESS in 7m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dddd9a70695e4b37bb4c69f3c4e92645 : SUCCESS in 7m 23s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/99ccb12b0451416c9da2ee736c2848f2 : SUCCESS in 9m 29s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/76f1f6134757474d8b5f779829fed621 : SUCCESS in 4m 40s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/060d9d034c2a405a810a3330fb25cd0d : SUCCESS in 1h 06m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4b0479033234ef19205ddadb1036a4e : SUCCESS in 53m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c3fc9e39e23e1ea74b533e1e23524ef2b6e2e1c4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-26 06:25:00.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"351f48ce0d2d2eea69eef679702329dffec3f238","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-26 20:58:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks fine.","accounts_in_message":[],"_revision_number":1},{"id":"4755667d1150d631b6f5da6490879290086843e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 20:59:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f040a05f52cfd6321d31df3de259e82ce1ce9d38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:11:48.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"8e4aaa09d218a448d4b71392d231a7280cf95d29","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-27 05:32:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"082da751edca59515817e9d8db021b10e16c8edc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 06:58:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/63817e5c128e4e08ad1cfe4a58042776 : SUCCESS in 1h 25m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4656a66c01db49b5800a41832e29a5c2 : SUCCESS in 1h 16m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f24fd860b51e45e384d8191ed3db4bbe : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf80ce4e7cf46c897cd5c21012ec3d9 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3323f5ae2d064342b196a8348d01aafb : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9aa9918d040441b9bd7ef86963b1fb2 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deec8d03b0a94e25b4fd3730b894c70a : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3235ce5062024f268d3b4697e8a2acd9 : SUCCESS in 6m 57s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1ba609b0315f475782a83f64f93f8863 : SUCCESS in 6m 50s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3e391028fbdb48e9935847d099f02504 : SUCCESS in 4m 13s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f784d2e60eda492193a14889750c8c07 : SUCCESS in 1h 07m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ec05f85bcde49e4bacf6112be54ff6d : SUCCESS in 56m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7d98623887010436567075c04ee2ecac5ec09295","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 06:58:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5af6b0385abd6d7d115263fd96aa248f11b93af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 08:07:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e1e90ab116e085b9f2cf05ec45dd73ad1421391b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 08:07:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/be0d0fcb17934eaa839d89ab2e1fe21a : SUCCESS in 1h 06m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9467abd16de9493da9770f1e0d6e40e2 : SUCCESS in 59m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0593c007166a488d8676a985beae24db : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f84d88c8144ff3b03d03f2007aa0ee : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82a15aaf6b614fb1b3137f3b906bfbf4 : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46fcd3ec14204f98808406ae88269d51 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5d9957fa8454a4aab46ff7e5fab9427 : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2c462bdc6e27448a98af7d07ce4fa24c : SUCCESS in 8m 29s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b67000c394d1444c9cd565887e2785bd : SUCCESS in 8m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a111f6bba144fd4a0914349cd41dbed : SUCCESS in 56m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5c78c09e058a14d87399b1f8b85578fe8c5c0b71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 08:09:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aead8a246d2c452bbecb29dd367d52b7 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56d01f25442a77aadc168e9fa943a5cb809f4ceb","revisions":{"56d01f25442a77aadc168e9fa943a5cb809f4ceb":{"kind":"REWORK","_number":1,"created":"2020-08-25 18:57:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/07/748007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/748007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/748007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/748007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/748007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/748007/1"}}},"commit":{"parents":[{"commit":"663e027e50533909833b7ef11d82b9bdcda86499","subject":"Flesh out FakeImage for extra_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/663e027e50533909833b7ef11d82b9bdcda86499"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-24 20:06:38.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-25 18:34:40.000000000","tz":-420},"subject":"Add image_set_property_atomic() helper","message":"Add image_set_property_atomic() helper\n\nThis adds a new DB API method to atomically create a property on an image\nin a way that we can be sure it is created once and only once for the\npurposes of exclusion of multiple threads.\n\nChange-Id: Ifdb711cb241ef13eccaa5ae29a234f2fe4a52eb8\nRelated-Bug: #1884596\n(cherry picked from commit 2a51843138e27071bf84269f6b2a601b3ba9978f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56d01f25442a77aadc168e9fa943a5cb809f4ceb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/56d01f25442a77aadc168e9fa943a5cb809f4ceb"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
