)]}'
{"id":"openstack%2Fglance~748009","triplet_id":"openstack%2Fglance~stable%2Fussuri~I136ef792d601326a67266e9ea9fcadd79ddba69e","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1884596","hashtags":[],"change_id":"I136ef792d601326a67266e9ea9fcadd79ddba69e","subject":"Heartbeat the actual work of the task","status":"MERGED","created":"2020-08-25 18:57:25.000000000","updated":"2020-08-27 22:53:57.000000000","submitted":"2020-08-27 22:53:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748009-1598568837033-46488824","meta_rev_id":"f92c6efd50195b0c771ed59ab20c184e8b4b81ad","_number":748009,"virtual_id_number":748009,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-27 22:53:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-26 21:08:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-26 06:25:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-26 21:08:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-26 06:25:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 06:25:26.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-26 21:08:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-27 22:53:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"598eff17a79825a33ac314d8072179345129947b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-25 18:57:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc6a5294c6a63605f25a99e242d98e650a00e72c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 20:21:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/53ec933559ef458caac3426298e076b3 : SUCCESS in 1h 11m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9998f7ed18f743d293b26a70999f3092 : SUCCESS in 1h 15m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/964315ed22b94f6284e54fa64a4cdfba : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9d44103c6d54abe9a24fb344137f2d0 : SUCCESS in 9m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e9c3e0d77e740d389f8ce6f81c999af : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f635f62a34d4a2183bb25a8cfb98990 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e81c01b04b1e4c968cdf0e4f8d426b04 : SUCCESS in 8m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d50b9e30be44ab699906addc0756bdf : SUCCESS in 8m 21s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e9a9ad5e06e84939bf81dcc1fa8bf730 : SUCCESS in 7m 34s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f0e3cda6becb49139c2f6c36ee8a32d0 : SUCCESS in 6m 03s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/def6c3e242be4e4dbb3428c1eaee2d41 : SUCCESS in 1h 01m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/325d33dcd408437aaf7d272b897d35eb : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":1},{"id":"256f1f1db4ec952239e612c44cc76a20286883af","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-26 06:25:26.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"2e2fed95e78dd58a746998aa148e782dc3660609","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-26 21:08:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good to me.","accounts_in_message":[],"_revision_number":1},{"id":"cc17ed58968498d85e8f0adc3b098f50d470cc9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 21:08:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ceea413fa88f2fd20928bcca393935d05ae2e5ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:11:49.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"f0010c2cb24873c61c66a49ded8ae4177a669ba6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-27 05:32:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc8a183c2fe322276f63d4972ec38e6bf8970acc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 06:57:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8bb5f79e883a436b9e2886cb1024f425 : SUCCESS in 1h 18m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b326a78795aa44c78c3a02cf5a997295 : SUCCESS in 1h 12m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d0c403c311a490c871573c92db56d01 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa45ccf4c5ae4d25ab9ac9335f46f9c3 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbcd08266d354dd99b7124b966c12b6d : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f252c09bb414d569798a3248a511d62 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55a6702810c34b88823369ed6429064e : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b31d7cdddc1744589b857cb1a2bcd5eb : SUCCESS in 6m 33s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/bc464f3264a4407faded52be8ca8eb42 : SUCCESS in 8m 19s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/34ec757a6b384c3db37c77cf2c0fbc7c : SUCCESS in 3m 49s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/83a2f14e47464e24926288ab4fbdc802 : SUCCESS in 1h 09m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc7bbc0b94334fe8a5c3a396042fe47e : SUCCESS in 58m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b0e13e6e24fa81feec1fb27618eefad8ba69fdba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 09:13:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc40d623667e22240b74f0d0028661698bb32e4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 22:53:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bca3b570efae45949ded7a5ac3bf51de : SUCCESS in 46m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ccfdadcff8ee4760baf3e1b1bad15dc4 : SUCCESS in 1h 07m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bed1ce025641413c9b07ad5db7a9311e : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d97cf56f92b74249b937831be220f8bb : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7650d40ed62649dea806b2aa1c04a468 : SUCCESS in 6m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67b02dbf3192485389dad9fc19c0071f : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6deb4d9784b4d47bedd2289a583c97c : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/430cf37cc86a4567b25700ab6b0a6bb0 : SUCCESS in 8m 30s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d28141cae2d84a2da04828b3157472e8 : SUCCESS in 9m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/846ee49a18e5467daeb7f7d7518cd161 : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":1},{"id":"791d5fd661d5801dd3659f4eb91e06104201c159","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 22:53:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"586ca78a230e98d286241080fb22d0b3ec2304b3","revisions":{"586ca78a230e98d286241080fb22d0b3ec2304b3":{"kind":"REWORK","_number":1,"created":"2020-08-25 18:57:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/09/748009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/748009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/748009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/748009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/748009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/748009/1"}}},"commit":{"parents":[{"commit":"825a0ec754f0e5c94819c9e22482841bf3b7fd16","subject":"Add image_delete_property_atomic() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/825a0ec754f0e5c94819c9e22482841bf3b7fd16"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 17:12:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-25 18:35:05.000000000","tz":-420},"subject":"Heartbeat the actual work of the task","message":"Heartbeat the actual work of the task\n\nThis introduces a CallbackIterator object that can sit in the data\npipeline when setting image data and provide periodic callbacks to\nallow recording status, checking for cancellation, etc. If the\ncallback raises an exception, that will bubble up to the read\noperation and cause an abort.\n\nThe heartbeat is timer-based, but cooperative, which means if the\ndata pipeline stalls, we will stop heartbeating. The goal is to\nprovide a callback about once per minute, although blocking on\nthe pipeline could make that less regular of course.\n\nChange-Id: I136ef792d601326a67266e9ea9fcadd79ddba69e\n(cherry picked from commit e49f23c04d784b5312d7df4556e2e5064555a84f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/586ca78a230e98d286241080fb22d0b3ec2304b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/586ca78a230e98d286241080fb22d0b3ec2304b3"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
