)]}'
{"id":"openstack%2Fglance~748010","triplet_id":"openstack%2Fglance~stable%2Fussuri~I8667c17813f6e701db98595b0b30df9e7b275294","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1884596","hashtags":[],"change_id":"I8667c17813f6e701db98595b0b30df9e7b275294","subject":"Update task message during import","status":"MERGED","created":"2020-08-25 18:57:25.000000000","updated":"2020-08-27 22:53:59.000000000","submitted":"2020-08-27 22:53:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748010-1598568839292-f4e8b138","meta_rev_id":"a2f50adf2ef1b2b909f348b41e948f4796fce484","_number":748010,"virtual_id_number":748010,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-27 22:53:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-26 21:08:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-26 06:25:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-26 21:08:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-26 06:25:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 06:25:35.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-26 21:08:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-27 22:53:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92d116c34a688548a25180062efd966673d91211","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-25 18:57:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4861ee5b5f484f8cdc9e6ac9bee29b6a59c5bc43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 20:15:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f48ad52543aa4e2b952a72439e337dfc : SUCCESS in 1h 10m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e5557f76a7124ba49fb3f82997e10fc9 : SUCCESS in 1h 03m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0970da4b6d0413982f683f99f9d9e8e : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06aa8db0d16b49d98ec8b43cde44a6eb : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5b58c2d5bf345f18a8d4a5e6bc489d2 : SUCCESS in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61200859d99943e198bcb0e75334b838 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c03f350f794f1484ebee1657d91ea5 : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aac15c3865944db39e06dd187f8ef05a : SUCCESS in 8m 21s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1681f1e7cdbd4ae1b0db2518c7ee9d52 : SUCCESS in 7m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/49b5ff2dd6c24e26ab42b7c384f79d69 : SUCCESS in 6m 35s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bf393b1d0c2a422b9deb4967f20b7fa1 : SUCCESS in 1h 05m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c3cbe2145504b08893decc04e7d23c9 : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":1},{"id":"eccf67f4e694a4f5416e2190ecb17d1d49d243cc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-26 06:25:35.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"81b5faf483134e751ebe3d4f41fdffec0f8fdec5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-26 21:08:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good.","accounts_in_message":[],"_revision_number":1},{"id":"99c62bae0974741eb518a841ce95f18f0136737d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 21:09:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a088d0067c0e3f4a7e136596cc375954aa15225d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:11:49.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"d487287c2444ccb605b0ef809003de37e3d047f3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-27 05:32:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8a836b2fc768949a8b08e16ff36d79463053ce6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 06:55:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/941fb884a8814b1b97b2864dd9fbd3b9 : SUCCESS in 1h 11m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/94796991e07248d89f7e734bb171e261 : SUCCESS in 1h 16m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/040cf418eb744c4b94ed4aa2bd7c5941 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ba64260ed324d5c8021fff3411bab23 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/345e802018e34ae4a3d5a1992134ba29 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20f8930106c34df099fcb0acd2e7f31a : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/340d432e10aa4c98975fcf9fcac7f4b0 : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b46d3ce6ece94b2794210e23cd646619 : SUCCESS in 6m 01s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d802a5ab996248b18384ddcb08efd572 : SUCCESS in 6m 39s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/97b244df5ae24021bed9754e6ee7e8ef : SUCCESS in 4m 07s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e8bd597afcfe4b119c1b16c9f0f6fa16 : SUCCESS in 1h 05m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9d790202f73416ab8cf80d785c4dd0d : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6ecff0d5740983e646ed5683c2e7fd5ef172a1a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 09:13:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac4e2fe3739c8ba13443bd5eb30bf0f37e807c41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 22:53:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8c7cdf428e3741b3b84d33a38b1e9c31 : SUCCESS in 46m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/28ca64b4ad6b416cb83500fd1b95f15b : SUCCESS in 1h 01m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa630f639ff34b928da117a57388baa7 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07cdccf4ff9343359a6eaea86ff4d5a1 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/007553969422438d9ef6a0d07a32efcc : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/848dcabd958a4290aa951fbfd4b95238 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/884944d6d79349908c480f560127aa0e : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dc0a9218b02a4cf7a8c2a4c8708ea51d : SUCCESS in 7m 19s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6ab8b465a1d544af999152a27b866f71 : SUCCESS in 7m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7b2b5fd69d14222b536efcd0d07c2d2 : SUCCESS in 36m 27s","accounts_in_message":[],"_revision_number":1},{"id":"969db529f6fbe49b28127adcac3bd8e054222803","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 22:53:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3da696e0a4c45b35cd9241ddc7c80c490b290ef2","revisions":{"3da696e0a4c45b35cd9241ddc7c80c490b290ef2":{"kind":"REWORK","_number":1,"created":"2020-08-25 18:57:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/10/748010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/10/748010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/748010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/748010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/748010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/10/748010/1"}}},"commit":{"parents":[{"commit":"586ca78a230e98d286241080fb22d0b3ec2304b3","subject":"Heartbeat the actual work of the task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/586ca78a230e98d286241080fb22d0b3ec2304b3"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 22:05:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-25 18:35:14.000000000","tz":-420},"subject":"Update task message during import","message":"Update task message during import\n\nThis updates the task.message field with details about the copy,\nwhich will also update the updated_at field on the task. This will\nfacilitate some rudimentary liveness checking from outside the task\nthread. Note that it also checks the state of our task, and will\nabort if it has been pushed out of \u0027processing\u0027 state externally.\nThis will be used in the following patch to faciliate import lock\nbusting behavior.\n\nChange-Id: I8667c17813f6e701db98595b0b30df9e7b275294\n(cherry picked from commit 77d9cfa66e3c2e70a1cc515e7504fd5dd946549a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3da696e0a4c45b35cd9241ddc7c80c490b290ef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3da696e0a4c45b35cd9241ddc7c80c490b290ef2"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
