)]}'
{"id":"openstack%2Fglance~748240","triplet_id":"openstack%2Fglance~stable%2Fussuri~I11c7edaefc96236d2757acfb70d9c338c0f51348","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1884596","hashtags":[],"change_id":"I11c7edaefc96236d2757acfb70d9c338c0f51348","subject":"Poll for final state on test_copy_image_revert_lifecycle()","status":"MERGED","created":"2020-08-26 13:47:35.000000000","updated":"2020-08-28 11:56:17.000000000","submitted":"2020-08-28 11:56:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748240-1598615777199-69a35627","meta_rev_id":"091066dd7281a9655ed92461f2285cf2d9e5d0de","_number":748240,"virtual_id_number":748240,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-28 11:56:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-26 21:12:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-27 05:33:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-27 05:33:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-27 05:33:11.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 21:12:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-27 05:33:11.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-28 11:56:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2bd9eca7e4e3e64f96f31d0e9d6b0f505016a79c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-26 13:47:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce41d600466118ab46480165cb37f1d00c2380e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 17:37:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a0fc49c90c34e8293c4c127319fdd0d : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aa76dd74417415c9d3f187fe3f655a1 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf6402ed08674ff7b737c135852efd22 : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/376f278868804f11a75aaea1fad8095b : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28bd0099da964a58be3fe493020caad0 : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f7a9acfc6fe1406292e417a1bf587f5a : SUCCESS in 9m 39s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/472bf38469c14d9c96ba0a8702886c10 : SUCCESS in 10m 12s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c0d3963836f243338b2243d5363b87f8 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"06761da9fe43f204658ce5b68e6cc93fd803c8a6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-26 21:12:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8f7eb86a6b75be101d36b17da9d24e42ce635466","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-27 05:33:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"8d29ebbc67b118208b7f44f53feb6ab5559f204e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 09:15:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d5814ea4a20ebc722516f23cc5106d3804e32ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 23:13:58.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"3bc82e6f82c384b8b17c1cadc9c219f5084e7128","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-28 05:51:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8eb176b71d25692d98fbc7bb90a0d753873432e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 06:06:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05656bb18b7c413e8b2708cd1e049b42 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6cbe43e01b74d519694e88e3f21e073 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e90a217a7043409b84cb2b60442aca9f : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a59ec0dabf59447bad931b22fb484d6a : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb920f42020e4bfe83a8c60c0573a3e2 : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/30dc60e31ffa458ab6d0bf5a2c5cf0d1 : SUCCESS in 9m 24s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/850bb16d18554d0ca97e00275af42fbf : SUCCESS in 7m 38s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/828e0f63feea48bf82c74e191205b270 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"41b2c8c5acdcc8776b44a77e0a4fef1c87d41706","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 07:26:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ce150e7cca84dd468c0e4cccaca29ceb0b00ec6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 11:56:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df31c7fd533848f08ee401964bfb0f55 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84ff42e762bb4fd8a0f516425cf39657 : SUCCESS in 9m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/546b0cf47d514fc49716fe1ae31b3b62 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7c4b77aa3ca4c3da60ea2fb51ef6557 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e6873b7155a4661a50c7db2455e7543 : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3185f8d1f404184bae48a6d5b689e02 : SUCCESS in 6m 48s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2d992db58434418cb6992f23d4e76a42 : SUCCESS in 8m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4cb539fa54cfe265443facc9c899168d2aea8c0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 11:56:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ebeb31e6367e39e81ea4ea62feeebb9f42f03109","revisions":{"ebeb31e6367e39e81ea4ea62feeebb9f42f03109":{"kind":"REWORK","_number":1,"created":"2020-08-26 13:47:35.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/40/748240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/40/748240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/748240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/748240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/748240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/40/748240/1"}}},"commit":{"parents":[{"commit":"541c0fd61e8e79492726c84a97a1186fbdf5c7a2","subject":"Fix non-deterministic copy_image_revert_lifecycle test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/541c0fd61e8e79492726c84a97a1186fbdf5c7a2"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-12 16:22:30.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-26 13:47:27.000000000","tz":-420},"subject":"Poll for final state on test_copy_image_revert_lifecycle()","message":"Poll for final state on test_copy_image_revert_lifecycle()\n\nThis test currently simulates failure by pre-deleting the store\ndirectory for \u0027file3\u0027 which is the second of a two-store import\noperation. The goal is to assert that the later failure reverts\nthe import of the earlier \u0027file2\u0027 store. However, the way the\npolling loop works is that we break out once \u0027file2\u0027 has completed,\nand then assume that \u0027file3\u0027 has already failed and reverted.\nThis is a race, and one we\u0027re losing consistently in CI.\n\nThis patch waits for the failure of \u0027file3\u0027 to be reported, as\nwell as the revert of \u0027file2\u0027 to occur before exiting the polling\nloop and checking the final state of things to ensure that the\nrevert has actually happened. This addresses the non-determinism\ninherent in the original test.\n\nChange-Id: I11c7edaefc96236d2757acfb70d9c338c0f51348\n(cherry picked from commit 6c96319eeb36274bed31466941dd83e7d9713cfd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebeb31e6367e39e81ea4ea62feeebb9f42f03109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebeb31e6367e39e81ea4ea62feeebb9f42f03109"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
