)]}'
{"id":"openstack%2Fglance~748390","triplet_id":"openstack%2Fglance~master~Id9b9260b03de83d41670ece4923a790cf88dbb13","project":"openstack/glance","branch":"master","topic":"ovn-devstack","hashtags":[],"change_id":"Id9b9260b03de83d41670ece4923a790cf88dbb13","subject":"DO NOT REVIEW: Test OVN devstack module","status":"ABANDONED","created":"2020-08-27 09:44:40.000000000","updated":"2021-05-20 08:40:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"985fcaa98aff64b236f8cff2ad4bb3da238e7b36","_number":748390,"virtual_id_number":748390,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2021-01-11 12:16:35.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"date":"2021-01-11 10:34:05.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-11 13:27:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-27 13:11:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-01 09:11:41.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-11 17:50:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"}],"messages":[{"id":"7d4d06891e9a05c021fbb1d2053c05301d371274","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-08-27 09:44:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4637fff15570dfef9c2d92fd2d559e7e59b8f2d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-27 09:45:04.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"56a6506f751f8fad1d6e82b7bfba8abf8509530c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 13:11:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/57688d2a9c6f4f63a8ce0c8afd0c3b37 : SUCCESS in 1h 01m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/930a032b70b54f2eb2da938ec5964f10 : SUCCESS in 1h 11m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79aa35f3122c418796607acc5cdb6389 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e878ae6665074aa98ede59f9945e8431 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6139838daa30479baaa716611c62a49f : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9e90a1cae9e4616abbb303b230f49ef : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc0a9c5c39e54a088c23991f0e5073f0 : SUCCESS in 9m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/74eca436015148feadcebc86fc463839 : TIMED_OUT in 31m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1fb05dc47939476f94348610aa430e17 : SUCCESS in 15m 15s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/08665feb4f7c417488eeab4ad2fd3f03 : SUCCESS in 5m 06s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b8f6d99a807d4b82b1504f6f8aa2603f : SUCCESS in 1h 01m 05s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/11488d2cbafe4f8e9f1d6cc05fe5268d : SUCCESS in 1h 35m 26s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/b75ccf18333447cda0e0ee22435e7ea8 : FAILURE in 10m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d46e1f4dc54c4601b664e9dee8374aab : SUCCESS in 1h 00m 19s","accounts_in_message":[],"_revision_number":1},{"id":"11bc1a0c603d903858c1784bee092987a07944f8","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-08-27 13:18:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0a2a57d447c86b3748c3c7426e290ea3032bda54","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-27 13:18:22.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"1870ff4b8bb163086a593105dc244205fa5904b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 15:46:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f28506652d054502a816da6de76886a7 : SUCCESS in 59m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/00d867c0e0774bea98af186199dc8d5e : SUCCESS in 1h 23m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b62f9b52c1fd43cf8f9706815a7430cc : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8878c3c3230f4fd394a37eae20a4d535 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51dafcd27c66430e9da3b2a634063795 : SUCCESS in 8m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c27d40241b654ccda273ead60374bca9 : FAILURE in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d7db49c73694ff3a51a1af51db21563 : SUCCESS in 7m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4414116f8bfa4e1eb3ac61ce6e4523b0 : SUCCESS in 9m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/54f3c4f0058a4e21b5ba4812d1381f0b : FAILURE in 5m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6b99611ef488473393089deae9faae0f : SUCCESS in 6m 39s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e061f136ef8b4fbf8357c6bee2b8345c : SUCCESS in 1h 12m 43s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/cc554fcb205443f8817afdb1d7e08e4d : POST_FAILURE in 10m 38s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/32d2b6b688c044809ea0acc4fa0d9869 : FAILURE in 1h 04m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64646721f3ba42538a1b9a908fcd9f8b : FAILURE in 10m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1b9f5f278d1f96d9bfde3150b08ecabc90aa63eb","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-09-01 09:11:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9baf5a15ccd3e94161b09f7e0bca7adb69542f5f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-01 09:11:41.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"74d9555468170632f833c59c0ce357705c761d41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 11:17:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/200bd51b24d04a1cbadd2a8a2f61eab4 : SUCCESS in 1h 05m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/699f0638ddce4127bff7936ad147c511 : SUCCESS in 1h 34m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6891be0e4abe4eaaafa72e1463feb2c3 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f4d557e48f5434aa9d2350cc4dc5f40 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f064e0caae1e4bbcbae982acb2c099ae : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff01b604999740e189243254b303da3d : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16fd2ccc638b475fb7efda8ad0fc6516 : SUCCESS in 8m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e44567bf64ad4688a0351696a4e4efaf : SUCCESS in 12m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5b85bb113cb44fa1a81db334b37b5f33 : SUCCESS in 13m 05s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/eaafb33819e34dd1b15e796df903a07d : SUCCESS in 5m 06s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8bc2222e26f742c28bea311e2940f0ac : SUCCESS in 1h 06m 13s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/74a92ecfd3ed44de9c95576f0f9427d1 : SUCCESS in 1h 05m 28s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/ba340857c67648ff8581992559b6b2c3 : SUCCESS in 1h 12m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16ed3bf3b99942019ce13a985ffd2cf2 : FAILURE in 1h 03m 32s","accounts_in_message":[],"_revision_number":1},{"id":"97aad03fae1fa4764de15dd8cec70030b1ccd01d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-01-11 10:34:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eb837d83ff737d23d3816f2ceabd808570499c48","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-11 12:16:35.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/748390/748390/1/check-not-vote/ext-glance-zuul/07ffc2c : FAILURE in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b54f3882f85a3d309d441699a690dec193e91617","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 13:27:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/736fb02457e54005bbdc1348764697bb : SUCCESS in 52m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/360e9feb3fd5488b90973d5d0c8a118d : SUCCESS in 1h 33m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8aa44ee53b644bca8dee86e4f5409d2 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e400da59d6a40f081947d248e0334b5 : SUCCESS in 8m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6706a04b8fdf45439747d46fd57b388b : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/574f20ba08de4e4a8c174d8ab380f5a5 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19af1e32604f4748ac3adc5e29245942 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74b17ff637a4423eae90548d79753ced : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/60861e82554b4585bbd06c717d273c38 : SUCCESS in 6m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cc47e24dc0ba48e5be0177ac255ffb4c : SUCCESS in 8m 52s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/cd7f4aff93e64f9a83ad2bbcf357115e : SUCCESS in 8m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3e20b5146d1d479d91c8558dcf778d13 : SUCCESS in 1h 58m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/22fc5999123b4ecbaf0da31d0d406a4d : SUCCESS in 1h 23m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/322bd4f0bd1940c2b4f35a1909b39a8f : SUCCESS in 1h 27m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/787ebda2869a4083bc2e6e74e331816f : SUCCESS in 1h 30m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7996bfaf4c6a42d0800862a44adb5045 : SUCCESS in 1h 18m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a09fc2e66a424b888c18ddf6487822e7 : SUCCESS in 1h 21m 29s","accounts_in_message":[],"_revision_number":1},{"id":"51b2fb316003700f54f0d2f4407c9d189a0923cd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-11 17:50:53.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"5446d226e919383b96faaa79f6aafe285ecbb76d","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 19:30:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- glance-tox-py38-glance_store-tips https://zuul.opendev.org/t/openstack/build/00661bf5ad924766bdaa57a115f4380f : SUCCESS in 5m 59s\n- glance-tox-py36-glance_store-tips https://zuul.opendev.org/t/openstack/build/e0604093dc9b4182a668957e4f3f0163 : SUCCESS in 6m 12s\n- glance-tox-functional-py38-glance_store-tips https://zuul.opendev.org/t/openstack/build/bf6b4f0a3dbd40c19d46b1686566565a : SUCCESS in 8m 07s\n- glance-tox-functional-py36-glance_store-tips https://zuul.opendev.org/t/openstack/build/7271f1255485444ca1de497050aae3eb : SUCCESS in 7m 34s\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d91ce74ba40e4ae386179360e812ca8d : SUCCESS in 42m 18s\n- grenade-multinode https://zuul.opendev.org/t/openstack/build/41ba68fe704540c5be525729b5581719 : SUCCESS in 1h 09m 27s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/38b1ae7e05054a9f847fc5f4f730470b : SUCCESS in 1h 29m 54s\n- tempest-full-py3-opensuse15 https://zuul.opendev.org/t/openstack/build/bb38a817dc9a482c8299c899f78873f2 : FAILURE in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"985fcaa98aff64b236f8cff2ad4bb3da238e7b36","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2021-05-20 08:40:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ccd08fed89a710bc6dbccb859fc1b484804812d","revisions":{"7ccd08fed89a710bc6dbccb859fc1b484804812d":{"kind":"REWORK","_number":1,"created":"2020-08-27 09:44:40.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/90/748390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/748390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/748390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/748390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/748390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/748390/1"}}},"commit":{"parents":[{"commit":"1b006c4f44e56d08d5ca1fb9090dc83990395d87","subject":"Add functional test for task status updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1b006c4f44e56d08d5ca1fb9090dc83990395d87"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-08-27 09:44:03.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2020-08-27 09:44:37.000000000","tz":60},"subject":"DO NOT REVIEW: Test OVN devstack module","message":"DO NOT REVIEW: Test OVN devstack module\n\nDepends-On: https://review.opendev.org/#/c/735097\nChange-Id: Id9b9260b03de83d41670ece4923a790cf88dbb13\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ccd08fed89a710bc6dbccb859fc1b484804812d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ccd08fed89a710bc6dbccb859fc1b484804812d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
