)]}'
{"id":"openstack%2Fglance~748450","triplet_id":"openstack%2Fglance~master~I9c734fabaabe78ea8f7e77d0aa2112ebe867ecb6","project":"openstack/glance","branch":"master","topic":"glance-import-job","hashtags":[],"change_id":"I9c734fabaabe78ea8f7e77d0aa2112ebe867ecb6","subject":"Run the nova-ceph-multistore job against glance","status":"MERGED","created":"2020-08-27 14:31:51.000000000","updated":"2020-09-18 15:56:28.000000000","submitted":"2020-09-18 15:50:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748450-1600444212480-2eec13d7","meta_rev_id":"2314e629e5983050482d7c32aee78cf7c4d1b25b","_number":748450,"virtual_id_number":748450,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-14 17:01:41.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 15:50:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-18 13:57:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-18 13:57:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-18 15:56:28.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-14 17:01:41.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-18 13:57:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-09-18 15:50:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0af5836bbb8ab784a4b6a140f7cf094dc19dac5a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-27 14:31:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ec3904bb16a15692867d7d61f0893a1a2c5d1a4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-27 14:32:32.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"dbc6583b7a45485074d5e559afd27d1866ae0b89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 17:27:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44728c2cc8fe414280a2be7a896fcc01 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90fbca30bbfd4c43a956c4f39eba57ad : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/175031d6d9224438a1f1a679bb565d9a : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbadee39eea14d0ba4747601170eca85 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c71d53a3f7264e47a9e532a75754f81c : SUCCESS in 7m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/acc8cd53a9ff411b9a51a17ef85e9ed5 : SUCCESS in 8m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b5dd5440e99543b8aa90aabf270d15f3 : FAILURE in 22m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/82570e87521d48aea9b77e930ccae972 : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":1},{"id":"609d44608aedd1627f1f413041692760de1aae28","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-31 14:04:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b3f6c9c9f56df61d30e3ec068dc1e319c7a428d6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-31 14:05:01.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"573fb587c511e2cc05f428db5fd39a5e2100a97a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 15:22:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a082499d2ee2402286cc50e7be95cadd : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59fcea5e5a6c404196dac27434b5bbca : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fa534f5b3514024a3321eae068e69b3 : SUCCESS in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ee5063b95824ce898afdda0e462b556 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a0c7c26061743a096532a24e787f0c9 : SUCCESS in 15m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5f40f11bcd1e4475bb1b741fdbd3dfad : SUCCESS in 9m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2de0cb425eab4372b3228da269d12954 : SUCCESS in 10m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3830a3d6e2e74e1fa5abcd1c4501b052 : FAILURE in 19m 43s","accounts_in_message":[],"_revision_number":1},{"id":"19bd1e3916b40ec50bf6eca117c13d1453894054","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-02 08:19:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5e8163b58dd94264f1d045bbd316f684bcb588f6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-02 08:19:45.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"d12d09f74c23af63aa5a4ec33a740ee6a2f57060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 11:02:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27156c5151d043a0a4d69967adb437d4 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc046f7f4e234fedb35b9ec5a14fbbf7 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fca2c2f0e74d4843b74a5fb3f3119646 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c96f87a12475499bba7d94f06223f6b7 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21b0e2d453c74dcf9f96f389f2547704 : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/95eb562a2a86468b993fc94ddadef9d1 : SUCCESS in 7m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/af8d545718d743c38f8db967f80b6f44 : TIMED_OUT in 31m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8d14ae0c3ae7472b8856165c873d7d85 : SUCCESS in 1h 07m 59s","accounts_in_message":[],"_revision_number":1},{"id":"5079a79f1b5408c24ea063feed451fc97360a3b2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-02 14:09:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ce10efc20f516e9c8928a07d6b98086246c19ada","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-02 14:10:19.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"c481008da447f3e6ff193133e611ef325cb4097a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 15:57:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cdd11ba01fb4cadb159400f26df7099 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66a8611545a24da4b3b5942458996230 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8399e8b9bc549febc3fbb35e4dea052 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d806bf114bb4668b7e4d7d0b3541b31 : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd841cb7cc74df4bb7dc9209437bd5d : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9d82acae94c1439f9829b9d49c20a769 : SUCCESS in 10m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/606991ec88c0493d8ec4449cb99b304c : SUCCESS in 11m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3512442dda1a480cbc70cb4884e2ad2e : SUCCESS in 1h 02m 48s","accounts_in_message":[],"_revision_number":1},{"id":"71b3d2b08a751bed5f2f39f7e3f809e067853891","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-03 05:39:08.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\nWhy depends on link with cinder devstack support patch?","accounts_in_message":[],"_revision_number":1},{"id":"db2d631d926d864807d88fcc144594e415dd2765","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-03 13:47:48.000000000","message":"Patch Set 1:\n\n\u003e Why depends on link with cinder devstack support patch?\n\nAs I said on that patch, I wanted to run it against that to make sure that the devstack patch wouldn\u0027t break the nova multistore support, just because it\u0027s swizzling things around in the glance driver. It\u0027s not an actual dependency for this, I just used it for the test. I don\u0027t think there\u0027s a major rush to land this, so it might be good to leave it to make sure we don\u0027t break that job with the devstack patch, but whatever you want.","accounts_in_message":[],"_revision_number":1},{"id":"970f3477c194a1544f5af3d83de157e828725007","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-14 17:01:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b3428077759d6f589dcddfdb85e40eb86b3d5c0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-14 17:01:41.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"fd6e63dd00d8765c16aca75729453789c24220da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 18:08:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8eae908b918c4d489825505e7cdb3b0f : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0b945db90fa406ca0d18b6e1efd0747 : SUCCESS in 9m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b1e5a0c83f74251a116c4acd04b52b1 : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38b07e1da4b14ec091b33c95dc21061d : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a085fdab5fed4b78ae557abc221854de : SUCCESS in 8m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/535c26268cec4eec816c9d0e1595b496 : SUCCESS in 7m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6caa8fee7c6149bd89954b84506a0cfd : SUCCESS in 8m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5c267d86358f4296847a36db2757d427 : SUCCESS in 1h 05m 18s","accounts_in_message":[],"_revision_number":2},{"id":"6d0ad1b528fd2443fca3b5f16504dd96302d2f19","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-18 13:57:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f679421cf28b0d90b9086cf28d930bc38dd34e2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 13:58:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"78bf20caeeea335d3e39c35948395dabe1b763f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 15:50:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17c1268aa97a40b88c317ed872adc62d : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9452433e2c74ff4b344c5178c6b049a : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef5105582eea44948381d324eb868f9a : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d0f6aea65ff48978bd7412ecaca02e8 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c5df85a86ed480b94fdf04ccd49cfc2 : SUCCESS in 7m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/78c8fcbfc2e54713a7d10cfff500073d : SUCCESS in 12m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6a80c7d2e2d74c5cb71a0fec0c5a6d37 : SUCCESS in 9m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ca21363f78c5465fb718feb2404945ed : SUCCESS in 1h 04m 22s","accounts_in_message":[],"_revision_number":2},{"id":"cb75359f1cdb9f3610207d896a861e0dd89ddb93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 15:50:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"36508d30d12b87b1f30bffce5dfd53f930699215","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 15:56:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef26218171794dfbaddc3d6c3b5a8949 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1a9b345840a3bff0f5508aad4d759400c6f7466e","revisions":{"b9ebdc0e72c1bef4f106efdac87d1186533c40cf":{"kind":"REWORK","_number":1,"created":"2020-08-27 14:31:51.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/50/748450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/748450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/748450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/748450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/748450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/748450/1"}}},"commit":{"parents":[{"commit":"c023445d051d0411998743df7fbdbefaaad1bd6c","subject":"Make our import-workflow job also convert images to raw","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c023445d051d0411998743df7fbdbefaaad1bd6c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-27 14:30:50.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-27 14:30:50.000000000","tz":-420},"subject":"Run the nova-ceph-multistore job against glance","message":"Run the nova-ceph-multistore job against glance\n\nNova has a job where both nova and glance are configured for multistore\nceph, and where the image gets automatically copied from the file to\nrbd stores on first use. Run that on glance to get the coverage for\nit as well.\n\nDepends-On: https://review.opendev.org/#/c/743800\nChange-Id: I9c734fabaabe78ea8f7e77d0aa2112ebe867ecb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b9ebdc0e72c1bef4f106efdac87d1186533c40cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b9ebdc0e72c1bef4f106efdac87d1186533c40cf"}]},"branch":"refs/heads/master"},"1a9b345840a3bff0f5508aad4d759400c6f7466e":{"kind":"REWORK","_number":2,"created":"2020-09-14 17:01:21.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/50/748450/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/748450/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/748450/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/748450/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/748450/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/748450/2"}}},"commit":{"parents":[{"commit":"e16d5c9ba3c687268eeef3df7a2832f173e42edd","subject":"Corrections in default value of all_stores_must_succeed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e16d5c9ba3c687268eeef3df7a2832f173e42edd"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-27 14:30:50.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-09-14 17:01:08.000000000","tz":-420},"subject":"Run the nova-ceph-multistore job against glance","message":"Run the nova-ceph-multistore job against glance\n\nNova has a job where both nova and glance are configured for multistore\nceph, and where the image gets automatically copied from the file to\nrbd stores on first use. Run that on glance to get the coverage for\nit as well.\n\nChange-Id: I9c734fabaabe78ea8f7e77d0aa2112ebe867ecb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a9b345840a3bff0f5508aad4d759400c6f7466e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a9b345840a3bff0f5508aad4d759400c6f7466e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
