)]}'
{"id":"openstack%2Fglance~748585","triplet_id":"openstack%2Fglance~master~I300a3f19a0dfacb23903ac3e92571855ed32cd83","project":"openstack/glance","branch":"master","topic":"print_param","hashtags":[],"change_id":"I300a3f19a0dfacb23903ac3e92571855ed32cd83","subject":"[Trivial]Add missing print parameters in log messages","status":"MERGED","created":"2020-08-28 08:14:34.000000000","updated":"2020-09-03 07:06:09.000000000","submitted":"2020-09-03 07:02:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748585-1599116555528-5b9a2423","meta_rev_id":"582649b2d970c517825d434ac1785b8c2afe8667","_number":748585,"virtual_id_number":748585,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-03 03:19:49.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-03 07:02:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"date":"2020-09-03 03:19:29.000000000","_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-01 06:54:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2020-09-02 08:17:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2020-09-02 16:03:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-01 05:20:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2020-09-02 16:03:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 05:20:00.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-09-01 06:54:47.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2020-09-02 08:17:23.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-09-02 16:03:48.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-09-03 03:19:49.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-03 07:02:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f33a926405f76f100c5b1fe35a90a909acc9c95","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-08-28 08:14:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c54f89c0c4ef3dc7f1c5e2d7d57b9ae09c8d672","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-28 08:14:56.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"7190991d887c9138b931a353328deeb1ecf81394","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 11:24:53.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ff985ffea9624093b336060fcc2b3817 : SUCCESS in 1h 03m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6293070b1a654559b05dc40ca02e3462 : SUCCESS in 1h 07m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31b407e48e564313bdabe61e5cd368ed : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb89a1b229f8451ab69da99670f827ad : FAILURE in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f1a5165bd7f4007bbbdb628c49fbbec : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ecf215dfcf941bca2821dec687d5991 : SUCCESS in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51a569a4845345299b7eb378d7152b5f : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b1945c27880244b2908b67af1e40bf1b : SUCCESS in 10m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b71dba1476184620a687a3636e72949a : SUCCESS in 9m 25s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d10033e181d648c2af1af5f9c44fe446 : SUCCESS in 5m 13s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/321ce3083d4643fcbd28b745f36dab77 : SUCCESS in 1h 05m 49s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/58a9d8d012c24bd49285494b50a847fc : SUCCESS in 1h 13m 38s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/e6327b80538044fc9e83aa5b09ac6fd4 : SUCCESS in 1h 08m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9753b85261bd4d3a8d1b913a6da2823f : SUCCESS in 53m 21s","accounts_in_message":[],"_revision_number":1},{"id":"6be00e386b17fc012c02ed1dce49548110a22128","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-08-31 03:30:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"516973ea6fcbbeaf936ea02db23dd971e1e1a275","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-31 03:31:10.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"b11e30355192c4bafe11c78a1143735c38165eb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 04:02:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f8645e181fd940eb9482a8e9f180f8b5 : FAILURE in 29m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0104a7992f2a40f38417413948268eaf : FAILURE in 31m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97dfdbdc52e94a88a47ef330e602026a : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf3d733558fb4db4a6a7c434059b59a7 : FAILURE in 28m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6166085eda443ad965f6a51f742d104 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39e91a10f38e4138960e5d0c27688b78 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e9b512ce3ee4e7d823cfe386eb59cad : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0e2eb74ca12f4d39a1768c8c857ba093 : SUCCESS in 7m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/72d7efa879da4187aae94935ee70d7e1 : SUCCESS in 7m 53s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/754868629f0a47c6aaea1e912f765b2d : SUCCESS in 5m 18s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6e3c60293c6d4517be13657e00384eb8 : FAILURE in 15m 21s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/88b05e78efce4d598afc4fd586112550 : POST_FAILURE in 16m 12s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/e27de00906e7427f95a42c0d25ac8258 : FAILURE in 14m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f731685465448f7afcb9abd4cc648cc : FAILURE in 13m 59s","accounts_in_message":[],"_revision_number":2},{"id":"a77eabb42a19f4258cc949c43ce19cfa69881200","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-08-31 08:42:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d9da628e10ca0fa21270974177c8748d193eb6fa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-31 08:42:28.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"88250a6a9a5b68be6b3def70e4b00da407e41b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 09:07:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/c110554f7bdb4e0da28e2590dec74411 : FAILURE in 18m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c3b372a20f74092a94a7f7c36cfb8df : FAILURE in 18m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e098d7e1acf04958a1e485352f541c82 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48e0a97fa54b4f0899c0ec73fbe0d183 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a6839605e764da19fc37aecc31942e1 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d59b95d44e84726a72ec6197c4bb995 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c867b3e8e8e8425db65772b761b842b1 : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ab0734a8b4c34b219efc346dd924d1c6 : SUCCESS in 7m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dc3c94b3aaf74ebfaeee431ddca4f7a1 : SUCCESS in 7m 27s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bbbc35c4194a49378ddf663a94525956 : SUCCESS in 5m 13s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3f3bf9a9861347cdb3983a92c3e9e357 : FAILURE in 19m 37s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/f696274eac514786a8da692e3822ea7f : POST_FAILURE in 17m 41s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/d8b003c2425b4d05a9be2ef4623d36d7 : FAILURE in 18m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d84eadad1a38415fa8f9ddbef4b87c81 : FAILURE in 18m 47s","accounts_in_message":[],"_revision_number":2},{"id":"355611868ee764b7d0e374a943def6a312f168f3","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-08-31 09:18:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5b52878daec510d62e6f04da8061c25729425687","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-31 09:18:31.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"94478e6f132aa468e7b375f6ae9c2a992db23382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 09:43:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/25c117b9c92b462a80b2dab2f306eb53 : FAILURE in 19m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01d2c4ae4e22447f8b88250ab8caece2 : FAILURE in 14m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed88fb81a3c44e07a9e2a87529b2af4c : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04e2daf36efa4429a6a446e692555f51 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14349c4635b647c6be5308314429ee84 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fee0046a2517430da7228cb609bfad07 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05edbf5e21b245afb02317759245152a : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/810eba19b09149a2899150057329d8ef : SUCCESS in 7m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/53c103b05bca4bbf9a1a8cbd87e6cb5f : SUCCESS in 8m 16s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9f4c681244c646318b127a265d0c8f4b : SUCCESS in 5m 02s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/039185307ad240b9a745473de1f82615 : FAILURE in 15m 20s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/98512efef44e434d8ce6adfe4b68d679 : POST_FAILURE in 18m 43s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/9d288ec0a5cb40679d89fb544dabe1ac : FAILURE in 14m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c7e50d7be5d43f3ba8c33728c6919d7 : FAILURE in 14m 42s","accounts_in_message":[],"_revision_number":2},{"id":"846f798a763ba7840644b044300d1f60b064ddaf","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-09-01 03:52:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"99c67b4723f59f4021ed53984fba42ad1aece4bb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-01 03:52:59.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"ef0ea34677b592efd8997a2ff9d1330b5e1e317b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 05:08:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/13bc0a2b61b94461b321020e817687b1 : SUCCESS in 41m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/50f5480f639b48a3bde2520b863c10a1 : SUCCESS in 46m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec9c1f77ca654af59ee405d40d6a45ca : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6bb063850ab4ee49160d89466517fdc : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e62c465d7ec647e4b0b1c0d1291fa9c3 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cceae904fab46d1ad722fee1f7b8aa1 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e83fe06f5cf47349fb2347f566e0097 : SUCCESS in 4m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e8cff13208b40298d32293736c87d80 : SUCCESS in 6m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/37bf70052f674b8f8d9276a0e1f6a40e : SUCCESS in 6m 55s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/21ce9a3ee85e4f1b89bce5d36fa5b54e : SUCCESS in 3m 38s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/fe5ca1afc0144b99a221e9b10c0c572a : SUCCESS in 1h 09m 16s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/946ec013ee9e40e3b44084ba1aadad53 : SUCCESS in 1h 06m 35s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/4edb78a83fd94838a41c140f860b94f0 : SUCCESS in 48m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19481f2b2a214e45a8118cccee5f695d : SUCCESS in 54m 09s","accounts_in_message":[],"_revision_number":2},{"id":"b056998f2591aabbdcc3884e945d2dfc5cf64dfe","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-01 05:20:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"be225f76834e047b173459234e439f9709e82a76","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-01 06:54:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4efa2b4d0b642715682664c6b24a3e28e7f3f1a2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-02 08:17:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8b4c10e5c16c75566a6f605d7f695e574a8a3f60","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-02 16:03:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4bffb06e3ab55658357bd4b717336aa4cbeb8a52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 16:04:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ebf4a20bae31f5a1e55efad11fccb7b636849ba4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 23:14:40.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/c81ddb477b224a89a721ec282f3c8410 : SUCCESS in 1h 10m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/009b8a6515f64bdfb06b0726601f9a63 : FAILURE in 1h 31m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5932768e14dd4176802d0655d8d6494a : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdfcda75e2064722b68cbb9ea9977012 : SUCCESS in 9m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72b16acb2d9a47edaa346a1dc38e6977 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3975624575ef4377a9e6ad4193d97f3d : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d011c84688f24e32803d705a0027c869 : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d89abab496774fe49821f64cf737e341 : SUCCESS in 9m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b2eb12c4855e4603a7c7740bcf3a38aa : SUCCESS in 10m 01s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/a904fd7be3e94f968debb07a9dcc106a : SUCCESS in 1h 08m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b72cde39672148fd83185e42c4297828 : SUCCESS in 57m 49s","accounts_in_message":[],"_revision_number":2},{"id":"ff230a591515d37f1589e41f7108ac62f72c5379","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-09-03 03:19:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0ec47f9fceb7294494db27e6cd2d13ad88448ca5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-03 03:19:49.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"37b8a4ca9db54fab20db80a13ca97416990b3e51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 04:29:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d774d9ad98e54dcf997c27c3d9c3c5b3 : SUCCESS in 58m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17af32fbe0714ef6a7ca596dee5b9b9a : SUCCESS in 1h 08m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d074b9ef1dc481ea6b607265f24375e : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/248290dc2e784ebdafbb2a976eea6d29 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df0b6dcba12f4adda654255f9899c62b : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcd062f3d6cf48a1841808806c9e7e10 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d481584f8d9460ab2f898862d65dea4 : SUCCESS in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa799cfa8241475bbbfe305accdaab22 : SUCCESS in 12m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d7dbd348c1324527ab21e8c57d9b1309 : SUCCESS in 7m 12s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/23a1a62d1099405a95129b72e4c7dde6 : SUCCESS in 4m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/070b9ba193ff40e3bef8e773d570d81e : SUCCESS in 55m 09s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/2dbbf127c7d14a4d8acd263c3176951f : SUCCESS in 46m 35s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/7947d71c0e4d4d95a93fff7c7d215022 : SUCCESS in 53m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8130e67d730149579e8e1b0ddf84c8fd : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":2},{"id":"1e659ed529ae57cc4b52f42827fe4b77dc37d74a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 04:29:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b1cdf4f7c6ea65c37dd86fd1ca8ed29d20b127a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 07:02:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7945f00830b38f397bd22536879e7ac87689891b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 07:02:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/43d9d54ddba8421a893bc5445c37404a : SUCCESS in 1h 00m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/acc32d5e5be74c26b4015c17d46f3622 : SUCCESS in 1h 14m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/212491741f9346378f9c930285ef8795 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db0ccac03ba8467080529eed18b0fedb : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0de83a729be74dee98e08ce4f68b48cb : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32036b5c17a44db8aafed0c7a4578e98 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b5ac6355f0141108a51bc1718a334cc : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/26996b2581b64bedaff74471d56f338c : SUCCESS in 10m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cc126c1b34ab4b44be3cf24d5be78b84 : SUCCESS in 9m 38s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/d22073924068409aa4329fbe25cc0ffe : SUCCESS in 1h 07m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b01b2f3de2f24ab7b2557b70ea7dcfb7 : SUCCESS in 1h 05m 35s","accounts_in_message":[],"_revision_number":2},{"id":"7e1ea34a0c83e4f0e81de16167f12c8a65b650ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 07:06:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c75229fd6a62424ea07d2aecf446f0b7 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"179d111c1fb5924e1703ee399d36abfe8286dec3","revisions":{"47ee8a1a99264a6d34edeb039c83f2690ef960f6":{"kind":"REWORK","_number":1,"created":"2020-08-28 08:14:34.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/85/748585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/85/748585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/85/748585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/85/748585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/85/748585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/85/748585/1"}}},"commit":{"parents":[{"commit":"0260ab5922231d6f974c0b73d30db69876c9bf5a","subject":"Merge \"Disable wait_for_fork() kill aggression if expect_exit\u003dTrue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0260ab5922231d6f974c0b73d30db69876c9bf5a"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-08-28 07:58:40.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-08-28 08:06:47.000000000","tz":480},"subject":"[Trivial]Add missing print parameters in log messages","message":"[Trivial]Add missing print parameters in log messages\n\nThis is to add missing print parameters in log message.\n\nChange-Id: I300a3f19a0dfacb23903ac3e92571855ed32cd83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/47ee8a1a99264a6d34edeb039c83f2690ef960f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/47ee8a1a99264a6d34edeb039c83f2690ef960f6"}]},"branch":"refs/heads/master"},"179d111c1fb5924e1703ee399d36abfe8286dec3":{"kind":"REWORK","_number":2,"created":"2020-08-31 03:30:47.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/85/748585/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/85/748585/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/85/748585/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/85/748585/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/85/748585/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/85/748585/2"}}},"commit":{"parents":[{"commit":"0260ab5922231d6f974c0b73d30db69876c9bf5a","subject":"Merge \"Disable wait_for_fork() kill aggression if expect_exit\u003dTrue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0260ab5922231d6f974c0b73d30db69876c9bf5a"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-08-28 07:58:40.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-08-31 03:25:05.000000000","tz":480},"subject":"[Trivial]Add missing print parameters in log messages","message":"[Trivial]Add missing print parameters in log messages\n\nThis is to add missing print parameters in log message.\n\nChange-Id: I300a3f19a0dfacb23903ac3e92571855ed32cd83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/179d111c1fb5924e1703ee399d36abfe8286dec3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/179d111c1fb5924e1703ee399d36abfe8286dec3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
