)]}'
{"id":"openstack%2Fglance~749510","triplet_id":"openstack%2Fglance~stable%2Fussuri~Ib910274472346ce0c336cd1ead8370d5799d0b96","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1884596","hashtags":[],"change_id":"Ib910274472346ce0c336cd1ead8370d5799d0b96","subject":"Handle atomic image properties separately","status":"MERGED","created":"2020-09-02 14:15:19.000000000","updated":"2020-09-18 01:18:18.000000000","submitted":"2020-09-18 01:14:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749510-1600391685787-515a9e8b","meta_rev_id":"f5669d4e445b9b0907099db1369f231c600aa57f","_number":749510,"virtual_id_number":749510,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 01:14:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-17 21:08:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-03 15:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-17 21:08:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-03 15:02:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-02 14:51:47.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-09-17 21:08:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-18 01:14:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bbe4a7880766783bdb99ecab51a287912895dab4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-02 14:15:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de27f7ff11be5a9f57b7cdabd5bb4f182b8eafe4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-02 14:51:47.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\nimportant backport","accounts_in_message":[],"_revision_number":1},{"id":"263accce036f63f30214288c24233fba95471665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 17:07:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/7c8f16ce352c4301a73a7bba753d3534 : SUCCESS in 1h 07m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc03c5dc29b34fc6a465a854de707187 : SUCCESS in 58m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35c882d9a86c432caf52397f36f75bd0 : SUCCESS in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4178892fb6004a15948c58c6cb33dced : FAILURE in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06765b3198c949aa9dd5a68543728b92 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7babf74524f0497fa13a03d343f65aa3 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec291bf13cf2411789caa2339353440e : SUCCESS in 7m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d124a275416a498393c2f618907595b9 : SUCCESS in 7m 21s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6a04f6519cf64e82bd88a0cd3f5b70c3 : SUCCESS in 8m 18s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7e3434660752418dabdc3fffa714bf03 : SUCCESS in 4m 37s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/06d8a167911749d0be38c93b677963b8 : SUCCESS in 1h 34m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/595bc0099045491ab2431b227ecae058 : SUCCESS in 1h 06m 07s\n\nWarning:\n  Comments left for invalid file glance/tests/utils.py","accounts_in_message":[],"_revision_number":1},{"id":"854ddcc41c98577e31adc1248b46138e7445e0c9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-03 15:02:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a5a0c24a9ec3716e223c8463a0be353503b43efe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 18:22:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/63b01988c7894fccb9f3677104a78ecc : SUCCESS in 1h 16m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dfa30566dc82438bb50341fe46ca5f19 : SUCCESS in 48m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10332bfa11b241ca9986911ea4b90bf4 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cde309e13f24470b6e7fc569fcd2a90 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3cece4c3184404b94d3ee3533ce266f : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6aa9d851a5a457f91049c4fe7aca6a1 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6229f22a4ef34c95884b035dda30234b : SUCCESS in 10m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1f88daa31da241a2a33a2d52e360c868 : SUCCESS in 9m 47s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3c0f1bfbdfef41a49ae66b5d7f57dddc : SUCCESS in 7m 54s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/20d1d4b2e08a4c6f9948f597789568ee : SUCCESS in 6m 00s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f127ec24c884456ab0ce9fc292af0c20 : SUCCESS in 1h 00m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70097eab2c0643cf8835dbef590dfc31 : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":2},{"id":"ecd0fd00f3b66938a02fc86623529b93500de236","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-17 21:08:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nMerged to master, good to backport.","accounts_in_message":[],"_revision_number":2},{"id":"ed13ff96412830701bbf78166f4fd0eb6bb31cce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 21:08:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1bbc973c8b1bad94c49e19bde50b967598b1ef91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 01:14:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"17937f3ddc8f1ef520592f727c571796868019aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 01:14:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b1236226d8d24d4f8d58c351d93c82dc : SUCCESS in 1h 08m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ab26a0bb39f549c0a32a39fc8b6623e5 : SUCCESS in 1h 07m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5bc0c5d735b43ab8fd3e682bc1c798f : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac42dbf9bf344f0fb90d6e25bd2819ea : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1501dd9f5c344908a702006bacfa258b : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9303cbaba82742a3a44fa7903df332b4 : SUCCESS in 11m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6acb700a055145baa038a15381c5fb8d : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/663a4192e65f4093bc0bf3f7d23ea682 : SUCCESS in 6m 50s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/adc559eb11db4c31a10863498f9dc2d9 : SUCCESS in 8m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c9988234eb4463b8cd5aa3e9fa51a79 : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":2},{"id":"c9fb05dd0b116db345a4c50ce7257d5d72109562","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 01:18:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9571c265229d466bab02c6c3cd15df45 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bb5731854a23e0803da028865342c036ad01dd60","revisions":{"0dd9dd766891186d9b81fbe41b793e002cbb2b57":{"kind":"REWORK","_number":1,"created":"2020-09-02 14:15:19.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/10/749510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/10/749510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/749510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/749510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/749510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/10/749510/1"}}},"commit":{"parents":[{"commit":"b272ffe8a29012d3929706d34c60400ba502d0de","subject":"Move SynchronousAPIBase to a generalized location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b272ffe8a29012d3929706d34c60400ba502d0de"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-17 13:58:00.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-09-02 14:14:21.000000000","tz":-420},"subject":"Handle atomic image properties separately","message":"Handle atomic image properties separately\n\nThe image_update() code will clobber, revert, or update lock values\nin keys that we use as atomic properties. This adds an exclusion\nlist of properties that we handle specially and plumbs them down\nto image_update() so that they will be excluded from the\nadd/update/delete logic.\n\nChange-Id: Ib910274472346ce0c336cd1ead8370d5799d0b96\n(cherry picked from commit 26f0311b29b18589ad134a34d5abd8ff2f82cf47)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0dd9dd766891186d9b81fbe41b793e002cbb2b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0dd9dd766891186d9b81fbe41b793e002cbb2b57"}]},"branch":"refs/heads/stable/ussuri"},"bb5731854a23e0803da028865342c036ad01dd60":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-03 15:02:03.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/10/749510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/10/749510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/749510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/749510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/749510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/10/749510/2"}}},"commit":{"parents":[{"commit":"beece27e0b2a7009e84e8888d4b6b395567b9458","subject":"Move SynchronousAPIBase to a generalized location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/beece27e0b2a7009e84e8888d4b6b395567b9458"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-17 13:58:00.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-09-03 15:00:35.000000000","tz":-420},"subject":"Handle atomic image properties separately","message":"Handle atomic image properties separately\n\nThe image_update() code will clobber, revert, or update lock values\nin keys that we use as atomic properties. This adds an exclusion\nlist of properties that we handle specially and plumbs them down\nto image_update() so that they will be excluded from the\nadd/update/delete logic.\n\nChange-Id: Ib910274472346ce0c336cd1ead8370d5799d0b96\n(cherry picked from commit 26f0311b29b18589ad134a34d5abd8ff2f82cf47)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb5731854a23e0803da028865342c036ad01dd60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb5731854a23e0803da028865342c036ad01dd60"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
