)]}'
{"id":"openstack%2Fglance~749511","triplet_id":"openstack%2Fglance~stable%2Fussuri~I74baf53fac1c3e23f6dc743058165ecb39074626","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1884596","hashtags":[],"change_id":"I74baf53fac1c3e23f6dc743058165ecb39074626","subject":"Functional test enhancement for lock busting","status":"MERGED","created":"2020-09-02 14:15:19.000000000","updated":"2020-09-18 01:23:00.000000000","submitted":"2020-09-18 01:21:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749511-1600392106361-2f639b6f","meta_rev_id":"5df362885c622c4b07b6f114a1a1935ad5b343c3","_number":749511,"virtual_id_number":749511,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 01:21:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-17 21:08:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-03 15:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-17 21:08:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-03 15:02:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-02 14:58:58.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-09-17 21:08:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-18 01:21:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d6599e7631dccc740abd2842f1135c329fd116d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-02 14:15:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0b83f0eea539003cc763192ac2093b70d9751ec","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-02 14:58:58.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\nImportant backport!","accounts_in_message":[],"_revision_number":1},{"id":"04f504ac9fb9782de6ea8d5c051daf47f19e64f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 17:08:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/51007362bd05435d948daafdf57c56d4 : SUCCESS in 1h 21m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c47376c6165f4c06bbc0c6bc5bcdd08c : SUCCESS in 1h 05m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/037d38c82340441f89190a8196391940 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48c4a1b788b74580bf7f068a6c8ddedc : FAILURE in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92c51003c4f04f5c8cb8b55ed3c92e72 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/330bfc5e9ea147f79daccf7d1b901d80 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b5cb59459b74e45b39f1fbc6768e806 : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/262677b7c51641e1abd1999a0bf5b5c3 : SUCCESS in 8m 00s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1966e21ba73643239f6d19b069a75cd6 : SUCCESS in 11m 12s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4c204d97f35b4fce83649a64bd576914 : SUCCESS in 5m 40s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/859970d2571d4638976e06e6afc2278f : SUCCESS in 1h 04m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec793c9b28b44d8cbacee0c9fc158e6a : SUCCESS in 50m 48s\n\nWarning:\n  Comments left for invalid file glance/tests/utils.py","accounts_in_message":[],"_revision_number":1},{"id":"53fd610a6f5034527da14a94e7d8c3b4aedc7c60","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-03 15:02:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0907f581dd9fabadc0371b8db620f990d4fc60c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 18:24:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/effc759452cf4e4bb32b74a0645fa89f : SUCCESS in 1h 13m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4368067f2ea34ae39039006297e4cdd9 : SUCCESS in 1h 13m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e60d152b3f7848c6a305160db1338463 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff898dee52a2498fbd4ca059f9545167 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1838a8d62e34411db45fa7401166dc24 : SUCCESS in 7m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa38e8db16524a17ac765b75c58ad048 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c0f026be0241f98348baeab4ffbc2a : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b375e929be2b4a928aece9ff9633b906 : SUCCESS in 8m 18s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/7d682f10403e4a1abab3adf1f4ac0abf : SUCCESS in 10m 36s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/085e3c764b184d5f8ec17d7c90880a2a : SUCCESS in 5m 17s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5c31b433aaa14f70ac2f46caf3b8482e : SUCCESS in 1h 04m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d972a10dc204a71b4ac7511821e51f1 : SUCCESS in 53m 33s","accounts_in_message":[],"_revision_number":2},{"id":"488b4399fa5d961f1992bf6348571cd8c9df5cf7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-17 21:08:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a37872dc7c4f5cba03a84091fc463b855a7b9233","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 21:08:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8109cd9725bad2a32421ebbd127d27246655f01a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 01:21:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dc1042b6fa57b3716538d46cf8680f984d65026a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 01:21:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f574fe2a82184f65b285212a0ff2e934 : SUCCESS in 1h 09m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0040b03e40554224825797f81a86979f : SUCCESS in 1h 11m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b7b31da6ec84bf9b999673558fed017 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03c448bdca1a4cb1ae82b6afd3e26a50 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a6de88ed83b43ec8ccb631442e67512 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d97a2d183c8c4e85b97a933d1a345916 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce9f757fdb504e0c884388f9bb5bc878 : SUCCESS in 7m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9d373bca54f94d4daa9b9dc2d527a1fd : SUCCESS in 7m 57s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6d4140fa73b446ed8a45e0f9ce1e212d : SUCCESS in 9m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7e0cc13110740abab5da2dff2530ba7 : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":2},{"id":"6fd8b27103c3ca47e208d3b8d76aa26aaee24834","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 01:23:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55001d1664fa433380474a52bfbcec61 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6c8f1e811c4622a320ada3b6dadb59d8840fdbb6","revisions":{"ea9389911b5ee312d58c7366517c06773ad192b8":{"kind":"REWORK","_number":1,"created":"2020-09-02 14:15:19.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/11/749511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/749511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/749511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/749511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/749511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/749511/1"}}},"commit":{"parents":[{"commit":"0dd9dd766891186d9b81fbe41b793e002cbb2b57","subject":"Handle atomic image properties separately","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0dd9dd766891186d9b81fbe41b793e002cbb2b57"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-17 14:08:21.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-09-02 14:14:42.000000000","tz":-420},"subject":"Functional test enhancement for lock busting","message":"Functional test enhancement for lock busting\n\nThis enhances our test for the import lock-busting case to include\nfreeing the stuck import task and letting the new and old ones proceed\nto make sure that the end state looks like what we expect.\n\nNote that this includes another task lock check after exiting the\nImportAction context before we call image save. While writing these\ntests I determined that we can end up with the original task racing\nto update the image locations. If set_data() took a very long time,\ncaused our lock to be stolen and another task is running, when our\nset_data() finally finishes we may overwrite their newly-added\nlocation when we go to save our (now stale) list. Thus, the extra\ntask check (imperfectly) tries to avoid us doing anything else after\nour task lock is stolen.\n\nChange-Id: I74baf53fac1c3e23f6dc743058165ecb39074626\n(cherry picked from commit a86062c4924a19e9ab9d5a3412353bc0aeceaeab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea9389911b5ee312d58c7366517c06773ad192b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea9389911b5ee312d58c7366517c06773ad192b8"}]},"branch":"refs/heads/stable/ussuri"},"6c8f1e811c4622a320ada3b6dadb59d8840fdbb6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-03 15:02:03.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/11/749511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/749511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/749511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/749511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/749511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/749511/2"}}},"commit":{"parents":[{"commit":"bb5731854a23e0803da028865342c036ad01dd60","subject":"Handle atomic image properties separately","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb5731854a23e0803da028865342c036ad01dd60"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-17 14:08:21.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-09-03 15:00:35.000000000","tz":-420},"subject":"Functional test enhancement for lock busting","message":"Functional test enhancement for lock busting\n\nThis enhances our test for the import lock-busting case to include\nfreeing the stuck import task and letting the new and old ones proceed\nto make sure that the end state looks like what we expect.\n\nNote that this includes another task lock check after exiting the\nImportAction context before we call image save. While writing these\ntests I determined that we can end up with the original task racing\nto update the image locations. If set_data() took a very long time,\ncaused our lock to be stolen and another task is running, when our\nset_data() finally finishes we may overwrite their newly-added\nlocation when we go to save our (now stale) list. Thus, the extra\ntask check (imperfectly) tries to avoid us doing anything else after\nour task lock is stolen.\n\nChange-Id: I74baf53fac1c3e23f6dc743058165ecb39074626\n(cherry picked from commit a86062c4924a19e9ab9d5a3412353bc0aeceaeab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c8f1e811c4622a320ada3b6dadb59d8840fdbb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c8f1e811c4622a320ada3b6dadb59d8840fdbb6"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
