)]}'
{"id":"openstack%2Fglance~749512","triplet_id":"openstack%2Fglance~stable%2Fussuri~I0795e3835a500f20102e0c393a9a15f00672c87a","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1884596","hashtags":[],"change_id":"I0795e3835a500f20102e0c393a9a15f00672c87a","subject":"Add ImageLock to base flow checks","status":"MERGED","created":"2020-09-02 14:15:19.000000000","updated":"2020-09-18 21:42:07.000000000","submitted":"2020-09-18 21:40:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749512-1600465221999-276e733d","meta_rev_id":"baacdd2914375df67cbfb4a61ba688cd069fd8fd","_number":749512,"virtual_id_number":749512,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 21:40:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-18 18:56:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-03 15:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-18 18:56:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-03 15:02:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-02 14:59:25.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-09-18 18:56:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-18 21:40:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"10045bccaa089aa7b06709da3ebce55154dfe3c1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-02 14:15:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c300adfe9d9221395e98991f47187c6dc0b4a9c2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-02 14:59:25.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"6e24d9f9b23fa2da98777fb4774891bc792a0084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 16:10:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/719060fb3c0848788e4f2aba81e6faeb : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b02ffa941b3f4eaa956dabf34d275759 : FAILURE in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/774cb567bd5847f8acda72651e6a2d96 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58a003af1fd74ad397360e40346b386f : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8c15fa5d2f2410688ca062c754c7f46 : SUCCESS in 8m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4e9f5953a4364dba9db694196837e044 : SUCCESS in 8m 18s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1ade9d8e85b44915b9689be12ea8f132 : SUCCESS in 7m 27s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/03fcda356af646fbb574e8ed4c759d07 : SUCCESS in 4m 43s\n\nWarning:\n  Comments left for invalid file glance/tests/utils.py","accounts_in_message":[],"_revision_number":1},{"id":"cfc5c57a32672414371d5c6bbcb440a2a865b3dd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-03 15:02:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"eb31c8d336e1ec5860169983a62e28c47ed8359a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 17:20:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9032f39284b04c97b26d69a0393708ef : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0276e2633bb42a7ad7f6be8bb934675 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/775b82140cd44e44bd0d667e3d1a83d9 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/973bf4771c104c12b4c8a915f01fa61e : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/327036aa577e47fc8cde30159374e42f : SUCCESS in 9m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5abdae859514455e80f870beab43946b : SUCCESS in 7m 47s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d311cbe63394479e9ebed469d4175852 : SUCCESS in 11m 20s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4b0dfc3074f94b7fafb46e16dfc33504 : SUCCESS in 10m 01s","accounts_in_message":[],"_revision_number":2},{"id":"535f189fc6b13c3cff671091b31f52405295b8b7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-18 18:56:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"67bc335bf31de7b893c30ea61a1dbf504c1c9e3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 18:57:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2ceba64ec9aed3653fb992f9721e0d295ebf8fd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 21:40:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8eb767823fc14b04a977da05875aee57 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d822e132484c3ba1e704aa65cfdcf4 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/681a1173faf441a696c970c268fc798b : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9b84824657a4428967891658932657a : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d378471942645a2af816d59a0edf898 : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3573b806886343538c929fc15a08ea22 : SUCCESS in 8m 30s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8820e16b2d034b328593879cd9d5b1ff : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":2},{"id":"9d0e4281bf2467f9930da804257bf72e3a4ef682","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 21:40:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1af5829374406039e36e46c2221f96f815e1bebb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 21:42:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7751688dc1547c4a20e0a6855fd5502 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"03a11dbbfb4c1f60946712413db449f8b52a4503","revisions":{"0bd0eca1c440e47d71a46c26133630decd477f02":{"kind":"REWORK","_number":1,"created":"2020-09-02 14:15:19.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/749512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/749512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/749512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/749512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/749512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/749512/1"}}},"commit":{"parents":[{"commit":"ea9389911b5ee312d58c7366517c06773ad192b8","subject":"Functional test enhancement for lock busting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea9389911b5ee312d58c7366517c06773ad192b8"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-20 22:29:04.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-09-02 14:14:48.000000000","tz":-420},"subject":"Add ImageLock to base flow checks","message":"Add ImageLock to base flow checks\n\nThis adds the ImageLock flow atom to the test_async flow assertions to\nmake sure it gets included. Also changed are the magic numbers for\nexpected flow lengths, per feedback on the\nIcb3c1d27e9a514d96fca7c1d824fd2183f69d8b3 review.\n\nChange-Id: I0795e3835a500f20102e0c393a9a15f00672c87a\n(cherry picked from commit 3636915dd4ce86b7dfeebaa5a1aa980a9ffaeb00)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bd0eca1c440e47d71a46c26133630decd477f02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bd0eca1c440e47d71a46c26133630decd477f02"}]},"branch":"refs/heads/stable/ussuri"},"03a11dbbfb4c1f60946712413db449f8b52a4503":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-03 15:02:03.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/749512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/749512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/749512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/749512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/749512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/749512/2"}}},"commit":{"parents":[{"commit":"6c8f1e811c4622a320ada3b6dadb59d8840fdbb6","subject":"Functional test enhancement for lock busting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c8f1e811c4622a320ada3b6dadb59d8840fdbb6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-20 22:29:04.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-09-03 15:00:35.000000000","tz":-420},"subject":"Add ImageLock to base flow checks","message":"Add ImageLock to base flow checks\n\nThis adds the ImageLock flow atom to the test_async flow assertions to\nmake sure it gets included. Also changed are the magic numbers for\nexpected flow lengths, per feedback on the\nIcb3c1d27e9a514d96fca7c1d824fd2183f69d8b3 review.\n\nChange-Id: I0795e3835a500f20102e0c393a9a15f00672c87a\n(cherry picked from commit 3636915dd4ce86b7dfeebaa5a1aa980a9ffaeb00)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03a11dbbfb4c1f60946712413db449f8b52a4503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03a11dbbfb4c1f60946712413db449f8b52a4503"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
