)]}'
{"id":"openstack%2Fglance~750144","triplet_id":"openstack%2Fglance~master~I2a3a99bd27db1c72d49b36b87e073e0b97fc874d","project":"openstack/glance","branch":"master","topic":"multiple-cinder-backend-support","hashtags":[],"change_id":"I2a3a99bd27db1c72d49b36b87e073e0b97fc874d","subject":"Add functional tests for cinder multiple store","status":"MERGED","created":"2020-09-07 09:44:56.000000000","updated":"2021-02-07 07:28:32.000000000","submitted":"2021-02-05 16:20:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"750144-multiple-cinder-backend-support","meta_rev_id":"301f1f622b1613e972935d96e2b9de4e1972061d","_number":750144,"virtual_id_number":750144,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-05 16:20:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-05 14:50:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-02-05 14:27:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-02-05 14:50:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-07 10:17:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-07 14:07:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-12-03 17:29:02.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-27 11:54:08.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-02-04 17:49:45.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2021-02-05 14:27:42.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-02-05 14:50:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"16c9e803b46845738098670c0142acf01e158180","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-07 09:44:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41197af97fd367cee3124cf17ba930ef6ccbeb02","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-07 09:49:01.000000000","message":"Patch Set 1:\n\nFunctional tests passing after glance_store changes (see depends-on)\n\nhttp://paste.openstack.org/show/797529/","accounts_in_message":[],"_revision_number":1},{"id":"0c14672c86e65d55c7a142f9836aa5015584207b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 09:57:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/86f005f4c9cd4a91bba86f8eee5d63e5 : SUCCESS in 4m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1c78df069b6d46bb84bd950ffbfb7fe2 : FAILURE in 3m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b8ce1d6962d4f0295e7bc7cb8278342 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c7647f2a81f4f66a3a353eb9fcf5515 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f256ee4f42b345cd89e760af6070599a : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a1ddd31c775411e80c6a18b5c4686c3 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96ed090fdf3a4c18ba52a6ddfc865219 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1e25b08063b4621baccb300f3002ba5 : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ea9a754927a44fa3b0fc86f32fdea321 : FAILURE in 7m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7ff979321b2943c4953780cad0faf140 : FAILURE in 10m 17s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3e21ba9b7051434183236177657b9bf4 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d3fdb0b2c8ac5c144fbd89f6ababeb6b305d1783","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-09-07 10:07:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"758e367a574bc8a6da6564a4b8f053264dec39cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 10:17:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b161590e123e4cd3a3daa596368d3d10 : SUCCESS in 5m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/59545980452e40cb88480d49248c4a1b : SUCCESS in 3m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17b0611aa09f49698769df413c5d4a6a : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11ec7fb2647949cb9dc93f189f4c5034 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1490ebece10b41c4951abcf966cd77e7 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0b65f30785d41f9b47627f416f6a6d4 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1669649987884cde8336f1dd67e234aa : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bb0a3feb0ef437d98eef11c595b3b62 : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fdc43f4423f844c1b416024701c1a929 : FAILURE in 8m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e42932fd4f9b49ea98f20308e56c7141 : FAILURE in 8m 51s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3f6f6c56d271427db6befbd463e9d79f : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":2},{"id":"ce3c6b1f17b91017c523dacb6f51e52fa7b40991","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-07 14:07:49.000000000","message":"Patch Set 2:\n\nAlso this change is required now as we have added functional test which required some test dependency packages needs to be added like python-cinderclient, os-brick and oslo.privsep and oslo.privsep has hard requirement to use project_id instead of tenant and user_id instead of user.","accounts_in_message":[],"_revision_number":2},{"id":"e536d89eed2dbcf995755dad3f7c0cb0d28fc90d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-12-03 17:28:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f17ab3abf4034b0dee2ffac033ec457085e987cf","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-03 17:29:02.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"c96e18889a1e2c45211972dae485acf6151e5237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 01:50:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6d6ce467cf4245b49df1642bebd6ba2e : SUCCESS in 5m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0b0fca52a7d74914b714ab75f03a1b3b : SUCCESS in 3m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b316758d63e466c903ccf54a0b10f90 : FAILURE in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53b099be9f0243878ad7e0bca6e694bf : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a892451799b2478687573e286c224847 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f6443b59e61453d93f8d6b6b4b7b767 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a21ae7f7fbc04de0b9c0c490004bc403 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8ec78c71f334a8cb474963644b2552e : SUCCESS in 8m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/90f61329daad4492a9a8541c6ad25230 : SUCCESS in 14m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d9be0f0ddae74a769d590194ab3a33dd : FAILURE in 31m 02s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/29b0f9b247204084880313e2a85b32c7 : SUCCESS in 4m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/65d84c51f567464f8aaeb375b0b8a60c : FAILURE in 1h 10m 04s","accounts_in_message":[],"_revision_number":2},{"id":"09195bb13d1be04282fa3423ea62dbe95fc17b95","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-12-04 13:54:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"48cbf9be082fd65276cf24318005be287d23ec03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 00:41:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/10b8d694018e4d4f9fd689bbe9928fc1 : SUCCESS in 4m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d7ff708663b847ed90d50fd0dc83c52e : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8372b21cebb84d2393961b88d67afb82 : FAILURE in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88df601ec5a04429ae1111c649929892 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6a42395047d4ab1a7d15992fe2c65da : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f380f149459041789852b507f108904a : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a01dad546803450a905fff9029fa1af2 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/631d46068c90438ba2b5b741e369d05c : SUCCESS in 5m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c98e18d71c0440309121b85de009eead : SUCCESS in 8m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/da901ffbf7214f1c9fc55f8619cde9fe : SUCCESS in 8m 48s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/88191f67aedf4576a2a601ed9aa8b7b7 : SUCCESS in 6m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4973613cff504f84ac42c4d139916260 : SUCCESS in 1h 12m 52s","accounts_in_message":[],"_revision_number":2},{"id":"4bcc11264033df8cd5df420cf80d6be42d499c66","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-05 09:24:01.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/750144/750144/2/check-not-vote/ext-glance-zuul/ed5067a : FAILURE in 6m 41s","accounts_in_message":[],"_revision_number":2},{"id":"41d0d5e0f978c8ac7f2a253127cd37eda2af971d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-01-21 05:52:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a32204b4d2ab0df09fcd10906c5cf4f63e1a1afc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 05:52:40.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"2171d391bdf8da5e3b8fe82d6bbcb8a1a9a52939","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-21 05:52:54.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"a7c2abf9584118e3b6b9316fd3270cb6c83285e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-01-21 05:55:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2ecca5d86fa4bf99c967588edfd4e527417fc8ad","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-21 06:01:11.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/750144/750144/3/check-not-vote/ext-glance-zuul/4509f06 : FAILURE in 4m 56s","accounts_in_message":[],"_revision_number":3},{"id":"8e72d80ad336a1a821220ed5d9079cb5974f0d56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 06:04:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e56420cbdb41462ea3031f2418dd6410 : SUCCESS in 4m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d17e3f617314461db9ab8547509a8031 : SUCCESS in 2m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07f99fd3430a446bb5f36c514e46b3cc : FAILURE in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aa2710ecc014363a1bcff975c74764f : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b54f73b1dbc84756bb8998bf79ebc693 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/434fc47850324cf5bf8f142c1b7bf138 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d5120c8943344c5a8f3ef35383bf6a2 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/302e15bd386743beb09d18026d16bf76 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25df595e64974c5788114703b68aaea9 : SUCCESS in 4m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e35a2fc4cfa14f738d12c0987e6fa237 : SUCCESS in 8m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/82941daeff0348e698060f04ab8d5470 : SUCCESS in 7m 39s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/aa92925d40db4a60a08471d7c0f5fb32 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":3},{"id":"9d481485f4cd48471e07cc3ed39571646a63835b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-27 11:54:08.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nNeed to check the lower-constraints (make sure everything in the l-c for os-brick 3.1.0 is updated for glance).","accounts_in_message":[],"_revision_number":3},{"id":"bb4ad89e771bdc26bb0c5bfddcd4eae9ead94c22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-01-27 13:02:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4e5fc7ef2c99bad9e20bd7bd334373d735184a3a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-01-27 13:06:07.000000000","message":"Patch Set 4:\n\n(1 comment)\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Need to check the lower-constraints (make sure everything in the l-c for os-brick 3.1.0 is updated for glance).\n\nThanks Brian,\nUpdated tenacity as per os-brick lower constraints.\nElse every os-brick requirement is equal or lower than the glance one so it should be good.","accounts_in_message":[],"_revision_number":4},{"id":"a5bd6678042f46bbf1910045a1b00b99b746b55d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 13:57:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b9b7fefe88324883b105ae041cc17024 : SUCCESS in 4m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/84e67bf0e1a649469a4a369958e13e8e : SUCCESS in 3m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97107db828414a8c90bc2dbf125a198e : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b6d4064d0f9406da693401895795b68 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fc4a8d9775c46ecb82c3bda82f5d7fb : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d48296134117460fb6b0fce4e63875e1 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06b7358603824b87852ae85a795d77f2 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/679395ca2d234042a8f535c99ed54e01 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4977e4a923b429fb6d926f05f3a1bad : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d284f04a0df04deea98b172404f2780d : SUCCESS in 7m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/53e1b5ea56104f32ad72a0162f5275b7 : SUCCESS in 8m 17s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f20032ff62f44243ac115b6017b267de : FAILURE in 4m 44s","accounts_in_message":[],"_revision_number":4},{"id":"e25f1fe6b7bd48009d37b3ae4b8d6d756553fcb4","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-01-28 10:48:11.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"8e3ce609ebb45681945baa2faae7538495d9b4c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 13:00:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/000a194f3ca84ffd8a9d4e063a58b203 : SUCCESS in 6m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2ad3c8c15fd649068d2a140bc6837950 : SUCCESS in 3m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f12884441af84fc08fd52307472f1662 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c79773ef1dd447feaa66e716f66b39cf : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ec6286a368b41d49fe871aa8fbcf357 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/091aa83fcec24f7da22a1f5eed74a7ed : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a98390a40dca41f7a94a37a5d036f8d3 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1baf55b516c6431b83e9ddb1088ea63a : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0beb334bfc30421da811ae90dc9338cd : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c4f9a050d5b248a38dd435eb994fe370 : SUCCESS in 8m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4c349fe4eb2846c8809fca4c02df65ae : SUCCESS in 8m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bf7c53c8fbe946c6a5bc0e4962291295 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":4},{"id":"98d6b4181048d9be375100a6e18ab812a471a118","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-01-28 13:43:25.000000000","message":"Patch Set 4: Code-Review+2\n\nLooks good to me!!","accounts_in_message":[],"_revision_number":4},{"id":"0eb77fcd09a75ea9d9f9988084f4a95e69fa9fc0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-28 19:38:25.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/750144/750144/4/check-not-vote/ext-glance-zuul/75cef4d : FAILURE in 5m 00s","accounts_in_message":[],"_revision_number":4},{"id":"f7ba3fc096c28d88dad54300ea9e77afbb5179b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-04 11:51:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7a307ea5eb5094121d26ad243bf9f068ea1674ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 13:22:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2bc1f7bad24c4e219ec4c0ab1e8a5f2d : SUCCESS in 5m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a54c3059c9b74e21b53345a13b04137f : SUCCESS in 3m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e25a4c61c4b743449eb66c4efb89231d : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd7aa1e276df41bb97b348b91f7f1cb8 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/097003f3964143139dcc3020099dc8f2 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a062f49adba042469041fe58abad3c0b : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff8d1359a22b4dd9a37edf3c262f131f : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04475f5ebe8e48ce9366a5ccf49e79e5 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a18e8882c90f4f64a3e5fe0c65e9389e : SUCCESS in 5m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ec484db2dd304d939523696f5588fc11 : TIMED_OUT in 31m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2bb7878fe7084598814c6a34bcef9791 : SUCCESS in 8m 42s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3b76a543871e4af3915473c0973390f9 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":5},{"id":"9d28d3ff607fa6687c39cf5db6009d059bb72a21","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-04 13:30:08.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0063b1c0dc846798a5d930768869b0165a93bf79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 15:06:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3076fe8b15924ac5a3f2bab93e0648a2 : SUCCESS in 4m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8a42f3df62ee453d913b932fc94d9d44 : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9dbb5750e35240fbb18e34ff67f704be : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15fea1a1d28340c2816e470d5eedc5d0 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e5083dc37cd4311ad62d70d65ee0b2f : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71f0813ac0dc4a559974298a457d41e1 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc2b555dac06487dbb5e382e20978869 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c07845bb04c4c71bf513848237f4e7e : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/062e3c45d10b40aeb4b853b138a3f643 : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/03ea03d563984a678579f955918968fa : SUCCESS in 9m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/420ba214939e4c23ae8751a1e5f8f627 : SUCCESS in 7m 22s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f97987db10394ac592df28bcc3d1b767 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":5},{"id":"6722eeafb62e20b21b00c005c2ad671b612d7e3f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-04 17:49:45.000000000","message":"Patch Set 5:\n\n(6 comments)\n\nRajat, sorry for ignoring this so long. This is good testing and thanks for doing it. I\u0027ve got a couple -1-worthy comments and some nits.\n\nWhen you revise this, please don\u0027t let me ignore it again. Ping me on IRC and I\u0027ll circle back.","accounts_in_message":[],"_revision_number":5},{"id":"c7377d1056a4b2b536d43b9ffc735aac4f2e4aec","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-05 12:25:19.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/750144/750144/5/check-not-vote/ext-glance-zuul/97d6d7c : FAILURE in 1m 06s","accounts_in_message":[],"_revision_number":5},{"id":"f8f5b359c252d14336b4a71c4970896b7d1c2d51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-05 13:11:01.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"08fbd46e357992a599ef021de8b5b60896006e41","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-05 13:14:05.000000000","message":"Patch Set 6:\n\n(5 comments)\n\nThanks for the review Dan!","accounts_in_message":[],"_revision_number":6},{"id":"3261007764f5a9b2fd183920c860f252e85dadbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 13:45:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e0f0f7f2c31d41ab9d13814e7e8caa2f : SUCCESS in 4m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d8209a4c66254c3a9ac07cb3da09a6ef : SUCCESS in 3m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d24c882b881d4221978d8016507d7783 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bc32298d01f4b62b67cc83d6ec205c3 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba2f3ce8266540eb94da6b19af3f1da5 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e50c26cdf864709bf05a293e952484d : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e7588b7dee743cc82dc285a77e6aa40 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7762e41ccf8a4d2d9551c2c2d8b9f425 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/824c9c17c5734fe8a072b15a4e806ebb : SUCCESS in 5m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ac997ac1216d4b7b91d91a53be681e4a : SUCCESS in 8m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1866e9df14ff453087cbe3e9880eb2e5 : SUCCESS in 8m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1b9a9adfd2d4486e8383328bfc4f47c2 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":6},{"id":"77f03597701a2bc39636695c03eb9abe32140577","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-05 14:27:42.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"bf6e2250da772df7380c9febf06af4ca374d0ac7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-05 14:50:50.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nLooks good to me.","accounts_in_message":[],"_revision_number":6},{"id":"dd91dd691e5715310dd53f6e26525f01f22043a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 14:51:12.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"af6e5609959d9d62e20ff59b4d5994f05fcb40ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 16:20:46.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9e644390fc0446939c0c7088c58e272e : SUCCESS in 4m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1c7a0c000d1e402a82d476f5bf8609b1 : SUCCESS in 2m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b12943d45dc749189944beab510fb222 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d0f9fc6366b41de8149c2ced849a371 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/433eda84762c472d863d5de93dae19e5 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecbf7819fb75465f971976472d34dd30 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b87201fbb3243ff85cf52fccdd5e113 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6eab53f028cf429c9a1deb83f00d6ef9 : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6706c2fc818e47a8b86ff4087d0b223b : SUCCESS in 8m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b0883c7dda9c4fcfa29a387a5fec1247 : SUCCESS in 9m 00s","accounts_in_message":[],"_revision_number":6},{"id":"2977ef56b1614c6ae37c6dfa6e71c62e0f20384f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 16:20:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"c2aa0a9f746cf8d8ad312598224ec6a4f6be837b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 16:23:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/d812164c25264233a0fcc46996caded3 : SUCCESS in 1m 06s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d1a8c1c892e46f1b4ae669c46b366fc : SUCCESS in 1m 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f8717b5af824ca28ab18c33f6f0bd5c : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":6},{"id":"301f1f622b1613e972935d96e2b9de4e1972061d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-07 07:28:32.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/750144/750144/6/check-not-vote/ext-glance-zuul/86acccd : FAILURE in 4m 28s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"09b924c9bcdfd4d750f285f60cfd02f5504908fb","revisions":{"a7bd346303fac9f702ac6491e26c794e62c552b1":{"kind":"REWORK","_number":1,"created":"2020-09-07 09:44:56.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/44/750144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/750144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/750144/1"}}},"commit":{"parents":[{"commit":"98a1e792c6cfa193d1977a80c56d7283ba00a013","subject":"Support cinder multiple stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98a1e792c6cfa193d1977a80c56d7283ba00a013"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-09-07 09:38:38.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-09-07 09:44:29.000000000","tz":0},"subject":"Add functional tests for cinder multiple store","message":"Add functional tests for cinder multiple store\n\nThis patch adds functional tests for cinder multiple stores[1] legacy\nimage migration and new image create.\nNOTE: This has been proposed separately as it has a dependency on a\nglance-store change[2] which will require a new release of glance-store\nto reflect changes on the gate.\n\n[1] https://review.opendev.org/#/c/748039\n[2] https://review.opendev.org/#/c/750131\n\nDepends-On: https://review.opendev.org/#/c/750131\nChange-Id: I2a3a99bd27db1c72d49b36b87e073e0b97fc874d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7bd346303fac9f702ac6491e26c794e62c552b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7bd346303fac9f702ac6491e26c794e62c552b1"}]},"branch":"refs/heads/master"},"0bd736c00c2dbba881b8985990f018efbd5600cd":{"kind":"REWORK","_number":2,"created":"2020-09-07 10:07:25.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/44/750144/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/750144/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/750144/2"}}},"commit":{"parents":[{"commit":"98a1e792c6cfa193d1977a80c56d7283ba00a013","subject":"Support cinder multiple stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98a1e792c6cfa193d1977a80c56d7283ba00a013"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-09-07 09:38:38.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-09-07 10:06:28.000000000","tz":0},"subject":"Add functional tests for cinder multiple store","message":"Add functional tests for cinder multiple store\n\nThis patch adds functional tests for cinder multiple stores[1] legacy\nimage migration and new image create.\nNOTE: This has been proposed separately as it has a dependency on a\nglance-store change[2] which will require a new release of glance-store\nto reflect changes on the gate.\n\n[1] https://review.opendev.org/#/c/748039\n[2] https://review.opendev.org/#/c/750131\n\nDepends-On: https://review.opendev.org/#/c/750131\nChange-Id: I2a3a99bd27db1c72d49b36b87e073e0b97fc874d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bd736c00c2dbba881b8985990f018efbd5600cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bd736c00c2dbba881b8985990f018efbd5600cd"}]},"branch":"refs/heads/master"},"3b2d886db1bd985eb5f547bf4c649d3a6b552da9":{"kind":"REWORK","_number":3,"created":"2021-01-21 05:55:47.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/44/750144/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/750144/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/750144/3"}}},"commit":{"parents":[{"commit":"a7f0e44699a4c74c631526664dbeb970601c003d","subject":"Merge \"Fix upgrade checks for sheepdog\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7f0e44699a4c74c631526664dbeb970601c003d"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-09-07 09:38:38.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-01-21 05:55:42.000000000","tz":-300},"subject":"Add functional tests for cinder multiple store","message":"Add functional tests for cinder multiple store\n\nThis patch adds functional tests for cinder multiple stores[1] legacy\nimage migration and new image create.\nNOTE: This has been proposed separately as it has a dependency on a\nglance-store change[2] which will require a new release of glance-store\nto reflect changes on the gate.\n\n[1] https://review.opendev.org/#/c/748039\n[2] https://review.opendev.org/#/c/750131\n\nDepends-On: https://review.opendev.org/#/c/750131\nChange-Id: I2a3a99bd27db1c72d49b36b87e073e0b97fc874d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b2d886db1bd985eb5f547bf4c649d3a6b552da9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b2d886db1bd985eb5f547bf4c649d3a6b552da9"}]},"branch":"refs/heads/master"},"735e654908b6c04fc4d3e795fc46fa61e82fdfac":{"kind":"REWORK","_number":4,"created":"2021-01-27 13:02:13.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/44/750144/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/750144/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/750144/4"}}},"commit":{"parents":[{"commit":"a7f0e44699a4c74c631526664dbeb970601c003d","subject":"Merge \"Fix upgrade checks for sheepdog\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7f0e44699a4c74c631526664dbeb970601c003d"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-09-07 09:38:38.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-01-27 13:02:07.000000000","tz":-300},"subject":"Add functional tests for cinder multiple store","message":"Add functional tests for cinder multiple store\n\nThis patch adds functional tests for cinder multiple stores[1] legacy\nimage migration and new image create.\nNOTE: This has been proposed separately as it has a dependency on a\nglance-store change[2] which will require a new release of glance-store\nto reflect changes on the gate.\n\n[1] https://review.opendev.org/#/c/748039\n[2] https://review.opendev.org/#/c/750131\n\nDepends-On: https://review.opendev.org/#/c/750131\nChange-Id: I2a3a99bd27db1c72d49b36b87e073e0b97fc874d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/735e654908b6c04fc4d3e795fc46fa61e82fdfac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/735e654908b6c04fc4d3e795fc46fa61e82fdfac"}]},"branch":"refs/heads/master"},"1aacda6cb2cef4fd1beefa6017eaea95c19bd073":{"kind":"REWORK","_number":5,"created":"2021-02-04 11:51:28.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/44/750144/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/750144/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/750144/5"}}},"commit":{"parents":[{"commit":"f2452863e74b563fdb4b1001af76ea492c63c484","subject":"Merge \"[goal] Deprecate the JSON formatted policy file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2452863e74b563fdb4b1001af76ea492c63c484"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-09-07 09:38:38.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-02-03 10:18:58.000000000","tz":-300},"subject":"Add functional tests for cinder multiple store","message":"Add functional tests for cinder multiple store\n\nThis patch adds functional tests for cinder multiple stores[1] legacy\nimage migration and new image create.\nNOTE: This has been proposed separately as it has a dependency on a\nglance-store change[2] which will require a new release of glance-store\nto reflect changes on the gate.\n\n[1] https://review.opendev.org/#/c/748039\n[2] https://review.opendev.org/#/c/750131\n\nDepends-On: https://review.opendev.org/#/c/750131\nChange-Id: I2a3a99bd27db1c72d49b36b87e073e0b97fc874d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1aacda6cb2cef4fd1beefa6017eaea95c19bd073"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1aacda6cb2cef4fd1beefa6017eaea95c19bd073"}]},"branch":"refs/heads/master"},"09b924c9bcdfd4d750f285f60cfd02f5504908fb":{"kind":"REWORK","_number":6,"created":"2021-02-05 13:11:01.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/44/750144/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/750144/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/750144/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/750144/6"}}},"commit":{"parents":[{"commit":"f2452863e74b563fdb4b1001af76ea492c63c484","subject":"Merge \"[goal] Deprecate the JSON formatted policy file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2452863e74b563fdb4b1001af76ea492c63c484"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-09-07 09:38:38.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-02-05 13:10:57.000000000","tz":-300},"subject":"Add functional tests for cinder multiple store","message":"Add functional tests for cinder multiple store\n\nThis patch adds functional tests for cinder multiple stores[1] legacy\nimage migration and new image create.\nNOTE: This has been proposed separately as it has a dependency on a\nglance-store change[2] which will require a new release of glance-store\nto reflect changes on the gate.\n\n[1] https://review.opendev.org/#/c/748039\n[2] https://review.opendev.org/#/c/750131\n\nDepends-On: https://review.opendev.org/#/c/750131\nChange-Id: I2a3a99bd27db1c72d49b36b87e073e0b97fc874d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/09b924c9bcdfd4d750f285f60cfd02f5504908fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/09b924c9bcdfd4d750f285f60cfd02f5504908fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
