)]}'
{"id":"openstack%2Fglance~750212","triplet_id":"openstack%2Fglance~master~I373467d2cdefb2301a949c9236f445dbbc641a2a","project":"openstack/glance","branch":"master","topic":"remove_osc_usage","hashtags":[],"change_id":"I373467d2cdefb2301a949c9236f445dbbc641a2a","subject":"Do not use OSC in infra playbook","status":"MERGED","created":"2020-09-07 20:31:24.000000000","updated":"2020-09-23 10:18:09.000000000","submitted":"2020-09-23 10:16:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750212-1600856168063-0e5654ba","meta_rev_id":"037106fa4c1a75dd6d62fbbb12fc6f85e2f7dba9","_number":750212,"virtual_id_number":750212,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-23 07:21:13.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-23 10:16:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-09-23 06:42:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":-1,"date":"2020-09-23 01:40:14.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-09-23 08:53:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 01:40:14.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-09-23 07:21:13.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-23 08:53:20.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-09-23 10:16:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"218b5ad671d5b0bf31ca38db48c4bf72ff60f3c7","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-09-07 20:31:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d60f2ee35461d348038686abf62f68ab4f115fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 21:56:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/361d91121a524d5ab8717cb1871dfa81 : SUCCESS in 1h 02m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/90458618d50f433e9eb3f166ecce4ab9 : SUCCESS in 1h 13m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2abdaf7045714e1cab1b4e59f7d1b3fe : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0b8c19a2d404c09936d687841766f46 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0560a90b9b9244ca87d956396f14742b : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/326220be66f3492b8aa509b930ac4409 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3fd3fce1d76408281a1e0cbe1621241 : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/673d4a734cba4d8f9b9c33e563a1c8ef : SUCCESS in 7m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f092ab40b58d4d9791336e93327cb767 : SUCCESS in 9m 56s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2dff43da3bc14b7eb6f4b5b1d813084f : SUCCESS in 4m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/08a5f837fb8f46a782ed3cf78562e86a : SUCCESS in 1h 01m 25s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/b070f7834abd4351b7195c36e8e057b0 : SUCCESS in 1h 18m 22s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/cb8db2d3923d40fdb90e1f48bd063c55 : SUCCESS in 1h 09m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dccb1789f4b456ab6877718a5ceef0a : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e7be843a092a25b94781dcfefd5f215d765f3afa","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-08 16:37:47.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks better from filtering and good use of limits.","accounts_in_message":[],"_revision_number":1},{"id":"15effb39daab87a643407f8b06aa8dafab720c7c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-17 14:43:03.000000000","message":"Patch Set 1: Code-Review-1\n\nI think we should be using osc for this, and I\u0027m not sure what optimization really matters in this case. It\u0027s run once after the tests are done. What\u0027s the gain here?","accounts_in_message":[],"_revision_number":1},{"id":"a4893d32cd30a70eb461cb88357fda4ee2e3f459","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-09-18 16:03:46.000000000","message":"Patch Set 1:\n\n\u003e I think we should be using osc for this,\n\nWell that\u0027s the part where we disagree. I don\u0027t see any reason why we should use osc, we do not support or endorse it for Glance, I see absolutely no reason why it should be used in something interfacing with glance-api and shipped in Glance repo.\n\n \u003e and I\u0027m not sure what\n \u003e optimization really matters in this case.\n\nUsing the filters and limits might make big difference, specially in environments where there is lots of images, on top of not needing to depend and load all the bloat osc brings in.\n\n \u003e It\u0027s run once after the\n \u003e tests are done. What\u0027s the gain here?\n\nIf the intended usage is just our infra ran tests, why it\u0027s not living under the glance/tests/ in the first place? Makes me wonder no less, why we should use osc, we don\u0027t support in the first place, for something we\u0027re relying our testing on.","accounts_in_message":[],"_revision_number":1},{"id":"e8bc8a01f6f83f2fa0930f456ef21ef196b1e8a1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-18 16:52:19.000000000","message":"Patch Set 1:\n\n\u003e Using the filters and limits might make big difference, specially\n \u003e in environments where there is lots of images, on top of not\n \u003e needing to depend and load all the bloat osc brings in.\n\nThere should be one image visible when this runs, only during a zuul job that actually calls it, of which there is precisely one.\n\n \u003e If the intended usage is just our infra ran tests, why it\u0027s not\n \u003e living under the glance/tests/ in the first place? Makes me wonder\n \u003e no less, why we should use osc, we don\u0027t support in the first\n \u003e place, for something we\u0027re relying our testing on.\n\nBecause this is where playbooks for infra stuff lives? For example:\n\nhttps://github.com/openstack/nova/tree/master/playbooks\n\nhttps://github.com/openstack/neutron/tree/master/playbooks\n\nhttps://github.com/openstack/cinder/tree/master/playbooks","accounts_in_message":[],"_revision_number":1},{"id":"0980b082f1da9204db61ebafa00f53c3dbeffffc","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-09-22 17:09:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"53c26826a26699b261f22427f2049b0307dab3ef","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-22 17:09:44.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"58e206e060ff8fdb2914344dd55a8c36e38ce947","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 23:31:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/970f646aad3e4cbea81d0a5897960f3e : SUCCESS in 1h 08m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b85361678b5d4399bff2512eadff9b21 : FAILURE in 1h 01m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55b04502417146388b472afbe7fa9eaf : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcea647186f84c7d97009254fd8d8659 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2c00c90af5a4c52a00cc631177808df : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e12285067eb47e495132abfc4758889 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/575e9689c26343168f3eb984bf388aaf : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/349381eac33d4ad4889e66cdb0222f3b : SUCCESS in 10m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/692cf03d00b6445dbec3ec41f1c2a06b : SUCCESS in 10m 08s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/716ce669084041bab9d52760f0faf121 : SUCCESS in 4m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b89cdcfceb804b1688b536cfe191140f : SUCCESS in 54m 37s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/f326155542094bf0a75eed75bc6eccee : SUCCESS in 1h 02m 01s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/e56bc9941e744d4c8c69e54fd6390532 : SUCCESS in 1h 06m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54ea3029f46f44ccb39be40be9d0dcda : SUCCESS in 49m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/073109d18c764b0389f7685941df9368 : SUCCESS in 1h 13m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4dfa6c06a287cab392c1966fde1a2b1498feb3ab","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-23 01:40:14.000000000","message":"Patch Set 2: Code-Review-1\n\nI still don\u0027t think we should merge this, but if we do, I\u0027d like to have my -1 on it :)","accounts_in_message":[],"_revision_number":2},{"id":"5e4c771a032bb7ec84c424f8e79d2c02ad15c585","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-23 03:02:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"69742c0d43f8db95345e26aa76356cb5739960a7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-23 03:02:55.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"eadfb0674a3fa7cc9493b39e75df110858052696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 04:17:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e9bd7f56850b46a88678c871be169c12 : SUCCESS in 1h 03m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d9708059aeef40e4b02c6c8ed83a3919 : SUCCESS in 1h 05m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4ddfbff68144c26a99e66f2437aa8f9 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aebdd0d090dc4527a34c6da5ece4c193 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a7795d2a4044932a540cd8e3169f8cb : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ed982c89e274ce285e5a1a08b93f557 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94269172b17b4a45923abfa39b3d4d74 : POST_FAILURE in 5m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/47acae206fbc46148654d63a4b8a4ce5 : SUCCESS in 7m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cf7acf459f8340aea6d043d2b76cb715 : SUCCESS in 6m 25s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/de7ee74c3f5a41bf8780a3f997524d91 : SUCCESS in 3m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/af6e8dc84606427aa6c9a34e2064d011 : SUCCESS in 1h 04m 40s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/c0112a31ccde4673ae83647833f9eb96 : SUCCESS in 1h 08m 58s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/59e29f1f54de43d880c44379bff83081 : POST_FAILURE in 58m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8feffad9232e47439d29d83e6da7c18e : POST_FAILURE in 37m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/15a61bcc604e4a698c895f6629f26d78 : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":2},{"id":"2a9e0c868d12847bb816f1ff4e4fa8f6f6567d76","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-23 05:52:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6290e0cada12666dbd52bc37d5ae7a29428cd5b3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-23 05:52:22.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"72acab7efd00d041c9199c7bd2b0c7c863f3fce8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-23 06:42:18.000000000","message":"Patch Set 2: Code-Review+2\n\nAs per our documentation we still does not support OSC, even though this is infra related stuff I still think we should use supported python-glanceclient only until OSC supports 100% of glance features.\n\nThank you!","accounts_in_message":[],"_revision_number":2},{"id":"e559b0243c869b43d686e209696e7144035ae8ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 07:15:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f6d6f7d76fe74cdbb7aeeb383d7b97f3 : SUCCESS in 1h 04m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de1da405b44f43078299ccfc198b1137 : FAILURE in 1h 17m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dd9d689375d404f8592b5b6d7a13caf : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bef1f20b375b4e5e9c107724d5f1a56a : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2769ba9a38e24ab6a001e4457d427c50 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bab154f9feb54597b88b3d1d05251e7c : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79ca95fcdf7940d8952a13f5115a5849 : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ac30e87a7894960a2a19699db2f9279 : SUCCESS in 7m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/af13dbba759246fb861680407d649abd : SUCCESS in 7m 22s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8abcbac0ccb8481689ae58bb08316866 : SUCCESS in 4m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/26d91f70f68a4d91a10b289057e9c537 : SUCCESS in 1h 05m 17s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/9fd377cbd488404289adb84e6a6a6950 : SUCCESS in 1h 12m 43s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/3934ee4f0de34212aa232fd19e49f5ee : SUCCESS in 1h 13m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59b978ffd99241128bc2cb3d739a8fc0 : SUCCESS in 1h 01m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a9a918e25bc94405acce83e63e8c295f : SUCCESS in 1h 03m 26s","accounts_in_message":[],"_revision_number":2},{"id":"79308920d6c1a16ae842d3ddb6ab88e0b2889de9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-23 07:20:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cc35969c2ca550d35a1e1a6bf23c0a06a16e5cba","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-23 07:21:13.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"1bfc8e2f9001986168257b5435ff223a53a08b5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 08:51:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/07840d75acd0422b8fed8d0fcba34e42 : SUCCESS in 54m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f3fc23f85a984353a5ee3a4e0e560043 : SUCCESS in 1h 06m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f717ca0dd48472b9a5b69a7c1d8193d : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5729df4feb04984afbb7500138b52db : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49101d0b90154c13acbd63ff3f558925 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af7be42ae5244239ad699aa164c2fe6d : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fffee164614a4818822bc4dc7eb83c57 : SUCCESS in 7m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8ee4be24c9ea473188b4f14fdad5bf1e : SUCCESS in 8m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4768648b795e409aa8cf8517c4150339 : SUCCESS in 10m 46s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/40d55b6ec9ef41089e0e91539a09fc90 : SUCCESS in 5m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ccf3e47dd80a4c64bfd8f42bc7ff955d : FAILURE in 1h 08m 38s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/6ab013575a1e435b9bce30083fb67cbb : SUCCESS in 1h 07m 37s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/7bab2c3035b940a0839926de226ee1c6 : SUCCESS in 1h 06m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1985954dc3a5459bb5be6a7215116671 : SUCCESS in 1h 12m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fbee50c1305b4151b8119c5b0e77e475 : SUCCESS in 1h 11m 22s","accounts_in_message":[],"_revision_number":2},{"id":"f3f0cd173ea7e2f79f353c4da4936e709b9369a3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-23 08:53:20.000000000","message":"Patch Set 2: Workflow+1\n\nConsistent with glance\u0027s core team decision as well as documentation saying not to use OSC until it is fully supported.","accounts_in_message":[],"_revision_number":2},{"id":"df632cc9c63bd3cfee6022999124d0d348fd9612","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 08:53:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5324a67705cfa32c3f5838a42eea5505907f5426","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 10:16:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a61a9044f17d4594a88071a41bf52b81 : SUCCESS in 1h 02m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/089cd2b6671745c896fb8559378204c9 : SUCCESS in 1h 13m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7e1eae193fe489696ddbeb6ebc4edac : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ab142631beb41808448f1ad0b9655c2 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/482e7f6499f1402693f2c96919c48146 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85a79c9ba3ab4ace820e9c30da9bc1aa : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/687e4a739f624af98f2f9a486bb8c93c : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5bfc1581e84b4cf5932149e1888dc699 : SUCCESS in 8m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/57b9f4b4c06e44749b14d7fe266a2b1c : SUCCESS in 7m 58s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/f7cf88af2e8e49d2bf60eff1c0cbac3a : SUCCESS in 1h 15m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/638824f1b2824364b86b2f6ce2e05dbb : SUCCESS in 1h 01m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf9fdd01561f433480a1991e050527c5 : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d131651592b649a425bb6d27f945083a0ca73838","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 10:16:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7394ea0e798bcb30a22a5229dbfb077baa1b3a6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 10:18:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055a94e450b84353ba7a28dad7a563dc : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"df0495676e84e89fe8eb84e3e4a26bba91a28186","revisions":{"c46c1faa1276b2a2f9dce648f2b35c492127096f":{"kind":"REWORK","_number":1,"created":"2020-09-07 20:31:24.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/12/750212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/750212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/750212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/750212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/750212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/750212/1"}}},"commit":{"parents":[{"commit":"5126ca0242ad0469bcd59c977f14d7d2ccd457e0","subject":"Merge \"[Trivial]Add missing print parameters in log messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5126ca0242ad0469bcd59c977f14d7d2ccd457e0"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-09-07 20:28:22.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-09-07 20:30:49.000000000","tz":60},"subject":"Optimize the image-import test post-check","message":"Optimize the image-import test post-check\n\nMinor tweaks to use glanceclient, it\u0027s property filtering\nand limits to optimize the post-check playbook.\n\nChange-Id: I373467d2cdefb2301a949c9236f445dbbc641a2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c46c1faa1276b2a2f9dce648f2b35c492127096f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c46c1faa1276b2a2f9dce648f2b35c492127096f"}]},"branch":"refs/heads/master"},"df0495676e84e89fe8eb84e3e4a26bba91a28186":{"kind":"REWORK","_number":2,"created":"2020-09-22 17:09:25.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/12/750212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/750212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/750212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/750212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/750212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/750212/2"}}},"commit":{"parents":[{"commit":"5126ca0242ad0469bcd59c977f14d7d2ccd457e0","subject":"Merge \"[Trivial]Add missing print parameters in log messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5126ca0242ad0469bcd59c977f14d7d2ccd457e0"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-09-07 20:28:22.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-09-22 17:07:39.000000000","tz":60},"subject":"Do not use OSC in infra playbook","message":"Do not use OSC in infra playbook\n\nUsing the supported glanceclient instead.\n\nChange-Id: I373467d2cdefb2301a949c9236f445dbbc641a2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/df0495676e84e89fe8eb84e3e4a26bba91a28186"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/df0495676e84e89fe8eb84e3e4a26bba91a28186"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
