)]}'
{"id":"openstack%2Fglance~751346","triplet_id":"openstack%2Fglance~master~Id7be6268ca0310e5ec67e22701d8cbf50ca245a3","project":"openstack/glance","branch":"master","topic":"glance-import-job","hashtags":[],"change_id":"Id7be6268ca0310e5ec67e22701d8cbf50ca245a3","subject":"DNM testing new tempest tests","status":"ABANDONED","created":"2020-09-11 16:17:49.000000000","updated":"2020-12-09 13:32:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"75176f9e4649b40507f17e12f4c89bd685d7c318","_number":751346,"virtual_id_number":751346,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-24 14:46:58.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"date":"2020-09-21 06:08:58.000000000","_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"date":"2020-09-24 14:46:38.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-18 20:07:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2020-09-18 18:00:33.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-18 15:43:59.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-09-18 20:07:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-21 06:08:58.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-09-24 14:46:38.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-09-24 14:46:58.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"285288c274f5034aa6bbce1a365390bcb59a8a4d","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-09-11 16:17:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81873ca24e2cb007a11aacf13b4810c7ee16c738","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 17:55:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/035ad9f68b5d4373b83b48d0b3a45e1a : SUCCESS in 1h 22m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9e509776822749d59e9767840b4f09e1 : SUCCESS in 1h 06m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4efb1a3862e5468d8f445d0a53a0f356 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30374a721ec84368a70424a8b7a34c42 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1121ba938f094807bf0c36f6d37af2e7 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb7df2fe8a65480da574cac19b2ade3f : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d304edec92b44dbb8334aedecc540072 : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3d522e4d18d64f7ba16b20bbadeb832f : SUCCESS in 8m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1e3b0cf434c5475085930359884047c6 : SUCCESS in 7m 13s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4f56a5e69cda439ca450d65e48b8c39d : SUCCESS in 5m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/27fd46a4592f4368b76c62b6361b1a85 : SUCCESS in 57m 19s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/6457d7c044c14f5cb9d71f4f125e9778 : SUCCESS in 1h 23m 14s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/2524ba9afaa14e68b289dfef4a38e29b : SUCCESS in 1h 25m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e80de4ee18047258825533e6c3a2dc0 : SUCCESS in 1h 13m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8526863affedef5f5be99f548e4a5f8f15ca74f5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-14 05:07:13.000000000","message":"Patch Set 1:\n\nSorry Martin, Sorry Martin,\n\nThe above mentioned job runs against single glance store, so these tests will be skipped there. These tests will be tested in\nhttps://review.opendev.org/748450 or https://review.opendev.org/750018\n\nOnce merged in master of glance","accounts_in_message":[],"_revision_number":1},{"id":"4acf06ab0b939f285dcf7517c9a71f06afb6ad95","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-18 15:43:59.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2bbc050714b5d78e572ddb581592dbaa455dbd64","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-18 15:44:24.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"e6b1116da3c7d2727c634fe3832e3e678096f58f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-18 18:00:33.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2ab634108b87a4b250b07b14f03938afdf9e26a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 20:07:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4b862dd19bb94e838c2c415b8a13af9c : SUCCESS in 57m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6847ad1b7e7643a98cc98bc303025bf8 : SUCCESS in 1h 01m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0af6c0b1dc040a58c528f1abd7c25e6 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/783cc91a26be4aeb9dfe86eeb1c12fd0 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e75dbc20c74c40ee85c4863c4090b739 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/442d050574fb45a887ae84c8a382c541 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6963be280054b5f98c1591eb739d264 : SUCCESS in 6m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb59d6790b30419f908d5dfd24146df7 : FAILURE in 10m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/80197a9a744a4f3db16c9b8c1b1f318d : SUCCESS in 11m 56s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ce56e8b469434f76863051d2afda7409 : SUCCESS in 5m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/07419ad05ff44a26b930556b5755a8e1 : SUCCESS in 1h 12m 05s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/80f29f30cafb4cbda9739fec0e5a98bf : SUCCESS in 1h 19m 59s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/1d63bb2c774344f686396282e178e0e8 : SUCCESS in 1h 24m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e3f761a6ba54fc4b269d5870a587877 : SUCCESS in 51m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/edcdd5784a894398b5cba1d402f5f7bc : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":3},{"id":"00978f9b26000f6d3459031f3dbaeef8311b1e16","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-21 06:08:58.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"67bb21f6464ad8f03a02c86095f97ddd93ef14a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 09:10:47.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/8b7832ed5f0149db90bfc6328290cc63 : SUCCESS in 48m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b52c18ffee6644faba39a7c5ba6a6ac4 : FAILURE in 1h 19m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a10d05158f74ad19d396d978c38b655 : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd54fc82f72e4cc2b4af7b80ba3b70b0 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44b38b98e02745ba973f067dbad492e8 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1951d4dbd0547f78f802e51ff6d0f5a : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a12168c0e64c4aaf923c8478529a78 : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1e9b2032958942b28329248f97354390 : SUCCESS in 9m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cb459c59328d4b8fba0aac60d2ed59ed : SUCCESS in 7m 48s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ba11a06249444315a7c169aa0418b861 : SUCCESS in 6m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cc8a85a653354370a17cb03e0025d99e : SUCCESS in 1h 00m 50s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/4e7284165c5f49a3a987899258dc51ee : SUCCESS in 1h 19m 24s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/c43fbab69e2b4ba297dcc4315ffaa608 : SUCCESS in 1h 05m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb486cd16e864e98b696e4380884c9ad : SUCCESS in 1h 02m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3c28f72e2af24bd5b9b205c3d5a8980b : SUCCESS in 50m 10s","accounts_in_message":[],"_revision_number":3},{"id":"b55017f7c0ca139ce95a327e56d44073ec29efcc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-24 14:46:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5eaaf32602a8bc50b40b303e9958199de82d596f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-24 14:46:58.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"f8c5d18e01917703323cd13aec88796495a2483a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 22:11:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/44ada1bda77b405d9637afe7c3f2100c : SUCCESS in 1h 01m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/78805114e96945bfaef7524874d72dfa : SUCCESS in 1h 30m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d908e9d31dee4b42bcacb88fd9b726c4 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af2e920c3a38433d87199e37f73b96ac : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/673827083f704f03903878abe44e9168 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/815eb1dc269e4f99ae720d76d295b7fc : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de0ac3f3dfa747949fd92c349c2e323b : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7bb9669471ed4d18852eaf585fc0e6a6 : SUCCESS in 8m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/37a8dd07a15148ad92bd962b67d0ee73 : SUCCESS in 7m 08s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8e785b912d4f4c86b25150558c5e30e3 : SUCCESS in 4m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/45b674ab9eaf4c3685521785e94a68d3 : SUCCESS in 1h 01m 13s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/c6cb599aa6564429aa9858cbe704c443 : FAILURE in 1h 01m 37s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/776eeb85b18c4ae1a532f3eee01e0576 : SUCCESS in 1h 10m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fb739e7f77c47e0820bc41ee1163033 : SUCCESS in 1h 12m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c4a45474ef63449ea3d290fe345416dc : SUCCESS in 1h 24m 20s","accounts_in_message":[],"_revision_number":3},{"id":"75176f9e4649b40507f17e12f4c89bd685d7c318","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-12-09 13:32:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"16f2336bfd654c3a364d4c3202c15aefc558215d","revisions":{"02ab6c7d054d0007497cf5d20a20cc311534ad36":{"kind":"REWORK","_number":1,"created":"2020-09-11 16:17:49.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/46/751346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/751346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/751346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/751346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/751346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/751346/1"}}},"commit":{"parents":[{"commit":"b1715c96ff57b03c02d0b6b6a5dcc5b005562f32","subject":"Merge \"[Docs] Cinder multiple stores for glance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1715c96ff57b03c02d0b6b6a5dcc5b005562f32"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-09-11 16:17:05.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-09-11 16:17:05.000000000","tz":0},"subject":"DNM testing new tempest tests","message":"DNM testing new tempest tests\n\nTesting new tempest tests added by this review:\nhttps://review.opendev.org/#/c/745712/\n\nDepends-On: https://review.opendev.org/#/c/745712/\nChange-Id: Id7be6268ca0310e5ec67e22701d8cbf50ca245a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02ab6c7d054d0007497cf5d20a20cc311534ad36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02ab6c7d054d0007497cf5d20a20cc311534ad36"}]},"branch":"refs/heads/master"},"134299863ac2e1b201ff90cf714dbc80eb431b0e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-18 15:43:59.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/46/751346/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/751346/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/751346/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/751346/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/751346/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/751346/2"}}},"commit":{"parents":[{"commit":"1a9b345840a3bff0f5508aad4d759400c6f7466e","subject":"Run the nova-ceph-multistore job against glance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a9b345840a3bff0f5508aad4d759400c6f7466e"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-09-11 16:17:05.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-09-18 15:43:39.000000000","tz":-420},"subject":"DNM testing new tempest tests","message":"DNM testing new tempest tests\n\nTesting new tempest tests added by this review:\nhttps://review.opendev.org/#/c/745712/\n\nDepends-On: https://review.opendev.org/#/c/745712/\nChange-Id: Id7be6268ca0310e5ec67e22701d8cbf50ca245a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/134299863ac2e1b201ff90cf714dbc80eb431b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/134299863ac2e1b201ff90cf714dbc80eb431b0e"}]},"branch":"refs/heads/master"},"16f2336bfd654c3a364d4c3202c15aefc558215d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-18 18:00:33.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/46/751346/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/751346/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/751346/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/751346/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/751346/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/751346/3"}}},"commit":{"parents":[{"commit":"ab151973bdf78c3a62ebfb5b102b221d85730c31","subject":"Merge \"Run the nova-ceph-multistore job against glance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab151973bdf78c3a62ebfb5b102b221d85730c31"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-09-11 16:17:05.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dms@danplanet.com","date":"2020-09-18 18:00:33.000000000","tz":0},"subject":"DNM testing new tempest tests","message":"DNM testing new tempest tests\n\nTesting new tempest tests added by this review:\nhttps://review.opendev.org/#/c/745712/\n\nDepends-On: https://review.opendev.org/#/c/745712/\nChange-Id: Id7be6268ca0310e5ec67e22701d8cbf50ca245a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/16f2336bfd654c3a364d4c3202c15aefc558215d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/16f2336bfd654c3a364d4c3202c15aefc558215d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
