)]}'
{"id":"openstack%2Fglance~752067","triplet_id":"openstack%2Fglance~master~Ie260486d795a6f4af1632f6f3708abc92fb47a3a","project":"openstack/glance","branch":"master","topic":"victoria-rc1-reno","hashtags":[],"change_id":"Ie260486d795a6f4af1632f6f3708abc92fb47a3a","subject":"Image import \"web-download\" check downloaded size","status":"MERGED","created":"2020-09-15 15:09:30.000000000","updated":"2020-09-22 23:19:38.000000000","submitted":"2020-09-22 23:17:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752067-1600816664912-72df3f54","meta_rev_id":"e0300c29e581b52e09c334d8bd77b121b3a81fc4","_number":752067,"virtual_id_number":752067,"owner":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-22 14:53:01.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-22 23:17:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-22 15:01:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2020-09-22 14:53:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-22 15:01:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},{"value":1,"date":"2020-09-22 23:19:38.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 06:08:29.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-09-22 14:53:01.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-22 14:53:42.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-09-22 15:01:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-22 23:17:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a6fdfba80a3276b8b667569ed65469157fb41ae","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-09-15 15:09:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b14717f2404cac7d0d2a5188ae775030477c724","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-15 15:09:50.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"4dfaaff6a4f0ab9684b4525acbb350980e74523a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 16:17:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a38f90320e3b41bf9d13f76da8dc7682 : POST_FAILURE in 27m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e7753d69ca89498394d293373d91c5c4 : SUCCESS in 1h 02m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa4c1f1b5e6a4eb997b63b2adba1b487 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1544ec9939f40db9235b1265454ea45 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cef612a8ca274bed9a37b550c3eec561 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ef80fb7dd8a4680a362cd718f43580b : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/851921b550e64230bbd4ef0d34ad7cd6 : FAILURE in 4m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e96115263cf748ca8cc563714b53f9ec : SUCCESS in 9m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/92ac925707664905bcc3e6bef40bd6cb : SUCCESS in 11m 22s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/49c5b456a6f84547a3f18ec706e7fefb : SUCCESS in 4m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d06a2635cfa1402485c62e6bbd3aa313 : FAILURE in 13m 26s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/637434c3e4ee4f459a2620ed421f0da6 : POST_FAILURE in 14m 09s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/28ad2b1c155f4c3fa7740d2ebd3967a6 : FAILURE in 11m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8beec43442834e4d86f77736ced17b0f : FAILURE in 11m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a3d574888f392e1969708c85de26c4f6a2a6a287","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-15 17:35:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"56e82dc4b8ffc3c4459afcdba2e7ce4f9be7f7d7","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-09-16 07:31:51.000000000","message":"Patch Set 1:\n\n(5 comments)\n\nThanks for the review","accounts_in_message":[],"_revision_number":1},{"id":"5aa1a6b384ffbac8d6e20f73a821708244eba614","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-09-16 07:32:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4199614fa3770dc4e0ed1945e6cb40886ee49ea3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-16 07:32:31.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"302e707bdd78e3430a4243638770b8b7351bf591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 10:01:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/27f2c8a7941546a5bb1a3c28850f612a : SUCCESS in 56m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3f0438339f3c4e7db043361730479583 : SUCCESS in 1h 00m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55ebc4127c7d4ca6a3ba5a692779d842 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c5fd3661172452eb97834d7189d5756 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fce6d5b3490f4eabb87b930ad4e00dc3 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47dbc21ecb924d2cb810f4bce37a2aa6 : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d33cad2afe6d4c5b9692572183bc0daf : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5b15f25887c54d48ba8c98e0c4c53e01 : SUCCESS in 10m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5551232c223f406bacd7b2ce8465b38a : SUCCESS in 8m 54s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b22c49e2c74c43749ab4a15444cee998 : SUCCESS in 6m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2cbd9c2bde944b29812f42e555965e75 : SUCCESS in 1h 01m 11s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/f7f5d66dae7c44f9b77d1d72ad3c230b : SUCCESS in 1h 09m 41s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/91e8f734343447f99f3816772421aaf1 : SUCCESS in 1h 28m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67b02ed712ac49cf9a8e414983114781 : SUCCESS in 56m 11s","accounts_in_message":[],"_revision_number":2},{"id":"e53b93ee224b1215954d7af94cf4fab3a97315eb","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-09-16 12:58:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1ff9f90b2a26dd11c9f1ebaa7e94b26003b403bb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-16 12:58:20.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"77f73acf6f22c6ab30d41681d411743b7cdf152d","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-09-16 13:03:33.000000000","message":"Patch Set 3:\n\nI\u0027ve added a new exception to be able to retrieve the path of the downloaded data in case of failure.\nI\u0027ll add the revert code in the change https://review.opendev.org/#/c/607680/ which implements the cleaning of the staging area in case of failure of the web download task.","accounts_in_message":[],"_revision_number":3},{"id":"01a74cf4d1cf44b5b148c62abb567e2a2846d745","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-16 14:38:37.000000000","message":"Patch Set 3: Code-Review-1\n\n(8 comments)\n\nThanks for the changes, a few more things. I\u0027m not practiced in he art of taskflow, so my comment there is just a comment.","accounts_in_message":[],"_revision_number":3},{"id":"0f3e29ccabfd69657271c00d242a93f0e66aea08","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-16 14:49:48.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nAgree with dansmith here, also it will be good if you create dependency of patch for https://review.opendev.org/#/c/607680/ with this one. This will help reviewers clear idea what is going on.","accounts_in_message":[],"_revision_number":3},{"id":"5ce315e4010a5b44ee152af05d304271136e71ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 14:51:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7e8e6894b5044a08a3213a3cdfdeaf6c : SUCCESS in 1h 08m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f4ae9dd448de4461a07fecee3158516a : SUCCESS in 1h 23m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fd148c729cd4394a4b5a3ce791dd08f : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd1547eb36e04dea96ba08606ef366b7 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2be299b5eac440f7b7ed840d700b55c7 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9897e3efca0549f58cb087d8f08a4469 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d7fe17a12f34780b590f8abb672a02d : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/687689623987435d91d7eaadd100cddf : SUCCESS in 7m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/24932c4f8c3045bcb656de7632e79d19 : SUCCESS in 9m 33s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d28d0bee6b6f441e95938c12b49f4c76 : SUCCESS in 6m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/841d105b1afb4954857a81a14aadb063 : SUCCESS in 1h 08m 07s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/5922b49b130b4e939868fc886b051eb4 : SUCCESS in 1h 15m 55s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/9ccd3fe95e944417a444f1438af3d2a3 : SUCCESS in 1h 10m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10c35de634fc422c98e40cad931285ce : SUCCESS in 59m 29s","accounts_in_message":[],"_revision_number":3},{"id":"1fdce7a2aab33f3783bf7631fd74c198832326f7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-16 14:53:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"757c89177840ff9dbf16a6ec87db90c4ebb6bae8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-16 15:10:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"76d861f20e346c9cc4e85805c3d38d2e3bd28458","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-09-17 07:42:50.000000000","message":"Patch Set 3:\n\nIn that case, maybe we can delete the data in staging only if the exception happens during the web-download task and not in another task of the import process?","accounts_in_message":[],"_revision_number":3},{"id":"a716c50d114f0155889b5993afb69c6837d5f2a0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-17 07:50:52.000000000","message":"Patch Set 3:\n\n\u003e In that case, maybe we can delete the data in staging only if the\n \u003e exception happens during the web-download task and not in another\n \u003e task of the import process?\n\nI think you should cleanup the staging data in revert method of webdownload plugin it self, just make sure to not to pass image path using exception, store it as task attribute suggested by dansmith.","accounts_in_message":[],"_revision_number":3},{"id":"68ae767ad7a8336d3ef44f0ec8a41409602c3d75","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-09-19 17:09:05.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"347765de2e721413572d5046a691f48574fc81b3","author":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"date":"2020-09-19 17:11:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"226627e7656357639e7736e4af0529ba642c129a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 18:39:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2a925159144c48f5a400009ca1aea221 : SUCCESS in 1h 01m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/313d95ed961641518c92a976ff8fa56f : SUCCESS in 1h 16m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/144ea75837594d8fb84101f991c0b3f6 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f5b0af7735e48b2b68ea20e5556f7f8 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d2b700a85b94a5d8d0fa9805b09dcd6 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c34ab78f50c40b099e419e4fab0aa7e : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b158495064742d4a6f51caf695f8221 : SUCCESS in 6m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0c2d828f12ba46bc84cb0f4549e77db9 : SUCCESS in 6m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/89e85687e4c8490d87828b01a140e0d3 : SUCCESS in 7m 22s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f04423d0552b428b9c497d1d2683076e : SUCCESS in 3m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a8e4337e050349f2b51da38aa6845004 : SUCCESS in 1h 21m 38s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/195f06aeb87a4631a943326aec19f0bd : SUCCESS in 1h 07m 25s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/3473365b32d54b02a0c2a623a87acdbc : SUCCESS in 1h 05m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a090544a9618462fafbb4dd8d481924f : SUCCESS in 57m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7b8ff45b43104b7ebcce4cf6102a0907 : SUCCESS in 57m 51s","accounts_in_message":[],"_revision_number":4},{"id":"03998aa2788ee54f5843a45fa3d29e0d21b1178e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-21 06:08:29.000000000","message":"Patch Set 4: Code-Review+1 Review-Priority+1\n\n(1 comment)\n\nLooks good to me, just one suggestion inline.\n\nWill vote +2 once dependency on cleaning staging data patch is set.","accounts_in_message":[],"_revision_number":4},{"id":"ef252832004396a9e72cd4855a743a9715c345ed","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-21 14:21:12.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nJust a couple nits for realism in the test. Looks a lot better now.","accounts_in_message":[],"_revision_number":4},{"id":"abbc4e110299c21c63a235287e1f4874bdac9454","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-22 14:52:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"95ac6ba6251469a6899ef0888907cab7a9e57d10","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-22 14:53:01.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"eef4894345d9cabec2c98bc1031a0ed449b9eafa","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-22 14:53:42.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"29186079e2fec0e8ec67dfaaf93fc9c3a195d33b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-22 15:01:18.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nLooks good to me. +W assuming zuul is going to be happy with this (and if it isn\u0027t, that will stop it from going anywhere).","accounts_in_message":[],"_revision_number":5},{"id":"abbde3f40184b8f091ab63f36a562b8af649fa73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 22:07:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5bff538168f746978f9d2def371387eb : SUCCESS in 1h 02m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3fdffe882c7443f39e964c0659186136 : SUCCESS in 1h 44m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b01fbf44389241e5ac80b34cea0a4ce0 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8eec3c1dc004450a6dd7a9dc02efe90 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3bca4ad97164cfe8c42512dfd069822 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2a9da9ac4604efb8a995f147ae8d2e1 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7dc3d4aca1c46fe95bb33af5afed451 : SUCCESS in 8m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5f7e958c13ad48b68417195c645d5847 : SUCCESS in 9m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/88f103a7bd37453190d6884bd86cecfa : SUCCESS in 7m 30s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6873c72f8cce42bcbc2071b11d8eab3a : SUCCESS in 5m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5168bfe8b25248bfb2af266d98a8d372 : SUCCESS in 1h 12m 29s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/1ba0878a69ab40bd9bb1c8c93485bb32 : SUCCESS in 1h 10m 16s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/4972eecc166d4b3c99be3fe4b0815251 : SUCCESS in 1h 01m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57caed56b0ef4a9eb2fdeaddd70a9d4e : SUCCESS in 49m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/07419d370f214a6f998fd5f12ce184a9 : SUCCESS in 1h 15m 32s","accounts_in_message":[],"_revision_number":5},{"id":"2c28bbcc5b6715ff61be2391c892ed70c22ba011","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 22:08:09.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b9562839b43328c68e5dcbebb1bec974e39891cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 23:17:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"b2c773488c134b42527fc5ca4a7888d565d0c695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 23:17:44.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/96f9beb875734469a32d87199eccace4 : SUCCESS in 1h 00m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/acee63557d92470fab5728200e175308 : SUCCESS in 1h 01m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a2e703b04084ce6a88c0d9997a9f532 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8866060b8f3439b8e1d87effd27985c : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b8360b2aa6f4049977ac747da1b9bf3 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39bd9186d5b74221b1d4aeb98c4d50bb : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3ce88414dac419a94abd5f67affb8b6 : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b5d3a94397a4ed28852e31a7c4efa21 : SUCCESS in 7m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/caae7b522bac4957816cec1214f8fa7e : SUCCESS in 8m 31s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/b4ee15f43c2b4dd1abea0216cf737f5d : SUCCESS in 1h 04m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6016acc1c94942b7af87b14300b3f6cc : SUCCESS in 51m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0daa0c8c8e4647e18418145be70bc9b9 : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":5},{"id":"c8e791a492db3bf3e2c0aa396926ba323ad0a825","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 23:19:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/518eccdec7a74f1b9bab73bc5f40fd61 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"68c202d38ba0648c8a71a7e4c3771ea6c6f60651","revisions":{"f6865f5617cf7e44b31e13a5868342b2e4ab99d9":{"kind":"REWORK","_number":1,"created":"2020-09-15 15:09:30.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/67/752067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/752067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/752067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/752067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/752067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/752067/1"}}},"commit":{"parents":[{"commit":"e16d5c9ba3c687268eeef3df7a2832f173e42edd","subject":"Corrections in default value of all_stores_must_succeed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e16d5c9ba3c687268eeef3df7a2832f173e42edd"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-09-15 13:19:43.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-09-15 14:03:48.000000000","tz":120},"subject":"Image import \"web-download\" check downloaded size","message":"Image import \"web-download\" check downloaded size\n\nIf the downloaded data size is different from the expected one, the\ntask \"web-download\" in the image import process will now fail.\n\nChange-Id: Ie260486d795a6f4af1632f6f3708abc92fb47a3a\nCloses-Bug: #1895663\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f6865f5617cf7e44b31e13a5868342b2e4ab99d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f6865f5617cf7e44b31e13a5868342b2e4ab99d9"}]},"branch":"refs/heads/master"},"e7b03a8e72e7f2459180fb9a4e8ae87baac48cae":{"kind":"REWORK","_number":2,"created":"2020-09-16 07:32:12.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/67/752067/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/752067/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/752067/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/752067/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/752067/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/752067/2"}}},"commit":{"parents":[{"commit":"e16d5c9ba3c687268eeef3df7a2832f173e42edd","subject":"Corrections in default value of all_stores_must_succeed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e16d5c9ba3c687268eeef3df7a2832f173e42edd"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-09-15 13:19:43.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-09-16 07:31:16.000000000","tz":120},"subject":"Image import \"web-download\" check downloaded size","message":"Image import \"web-download\" check downloaded size\n\nIf the downloaded data size is different from the expected one, the\ntask \"web-download\" in the image import process will now fail.\n\nChange-Id: Ie260486d795a6f4af1632f6f3708abc92fb47a3a\nCloses-Bug: #1895663\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e7b03a8e72e7f2459180fb9a4e8ae87baac48cae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e7b03a8e72e7f2459180fb9a4e8ae87baac48cae"}]},"branch":"refs/heads/master"},"c600b191ccc1f34c406bcff179d8cc9517ac66eb":{"kind":"REWORK","_number":3,"created":"2020-09-16 12:58:01.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/67/752067/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/752067/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/752067/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/752067/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/752067/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/752067/3"}}},"commit":{"parents":[{"commit":"e16d5c9ba3c687268eeef3df7a2832f173e42edd","subject":"Corrections in default value of all_stores_must_succeed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e16d5c9ba3c687268eeef3df7a2832f173e42edd"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-09-15 13:19:43.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-09-16 12:57:45.000000000","tz":120},"subject":"Image import \"web-download\" check downloaded size","message":"Image import \"web-download\" check downloaded size\n\nIf the downloaded data size is different from the expected one, the\ntask \"web-download\" in the image import process will now fail.\n\nChange-Id: Ie260486d795a6f4af1632f6f3708abc92fb47a3a\nCloses-Bug: #1895663\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c600b191ccc1f34c406bcff179d8cc9517ac66eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c600b191ccc1f34c406bcff179d8cc9517ac66eb"}]},"branch":"refs/heads/master"},"e261630cc498335528ea7487e47cd34e3908f54a":{"kind":"REWORK","_number":4,"created":"2020-09-19 17:11:24.000000000","uploader":{"_account_id":30054,"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","username":"yebinama"},"ref":"refs/changes/67/752067/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/752067/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/752067/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/752067/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/752067/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/752067/4"}}},"commit":{"parents":[{"commit":"e16d5c9ba3c687268eeef3df7a2832f173e42edd","subject":"Corrections in default value of all_stores_must_succeed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e16d5c9ba3c687268eeef3df7a2832f173e42edd"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-09-15 13:19:43.000000000","tz":120},"committer":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-09-19 17:10:59.000000000","tz":120},"subject":"Image import \"web-download\" check downloaded size","message":"Image import \"web-download\" check downloaded size\n\nIf the downloaded data size is different from the expected one, the\ntask \"web-download\" in the image import process will now fail.\n\nChange-Id: Ie260486d795a6f4af1632f6f3708abc92fb47a3a\nCloses-Bug: #1895663\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e261630cc498335528ea7487e47cd34e3908f54a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e261630cc498335528ea7487e47cd34e3908f54a"}]},"branch":"refs/heads/master"},"68c202d38ba0648c8a71a7e4c3771ea6c6f60651":{"kind":"REWORK","_number":5,"created":"2020-09-22 14:52:39.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/67/752067/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/752067/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/752067/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/752067/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/752067/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/752067/5"}}},"commit":{"parents":[{"commit":"ab151973bdf78c3a62ebfb5b102b221d85730c31","subject":"Merge \"Run the nova-ceph-multistore job against glance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab151973bdf78c3a62ebfb5b102b221d85730c31"}]}],"author":{"name":"Grégoire Unbekandt","email":"gregoire.unbekandt@gmail.com","date":"2020-09-15 13:19:43.000000000","tz":120},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-09-22 14:04:15.000000000","tz":0},"subject":"Image import \"web-download\" check downloaded size","message":"Image import \"web-download\" check downloaded size\n\nIf the downloaded data size is different from the expected one, the\ntask \"web-download\" in the image import process will now fail.\n\nChange-Id: Ie260486d795a6f4af1632f6f3708abc92fb47a3a\nCloses-Bug: #1895663\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/68c202d38ba0648c8a71a7e4c3771ea6c6f60651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/68c202d38ba0648c8a71a7e4c3771ea6c6f60651"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
