)]}'
{"id":"openstack%2Fglance~752392","triplet_id":"openstack%2Fglance~stable%2Fussuri~I53454c05710ebc45381315a20e10e627cbb65650","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1876419-stable/ussuri","hashtags":[],"change_id":"I53454c05710ebc45381315a20e10e627cbb65650","subject":"Fix a failure to parse json file","status":"MERGED","created":"2020-09-17 08:36:14.000000000","updated":"2020-09-18 07:17:46.000000000","submitted":"2020-09-18 07:15:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752392-1600413309781-5dd22bf2","meta_rev_id":"591f060a3dabb968cb0d9863ecd53cdbe10f2a39","_number":752392,"virtual_id_number":752392,"owner":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 07:15:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-17 17:59:28.000000000","_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-17 12:51:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2020-09-17 08:37:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":1,"date":"2020-09-17 13:40:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":2,"date":"2020-09-18 06:06:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2020-09-17 08:46:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":1,"date":"2020-09-18 06:06:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-09-17 08:36:14.000000000","_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-17 08:36:14.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2020-09-17 08:37:53.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-17 08:46:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2020-09-17 12:51:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-17 13:40:16.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2020-09-18 06:06:54.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-09-18 07:15:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df978438bc6602942157329e2aba251372829277","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2020-09-17 08:36:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8a88307f287915a123cc787a7b893860d846bb0b","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-17 08:37:53.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"99cc330d88335836f7070a02f9bfd6cfa772a258","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2020-09-17 08:46:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"41fe98facb16e24fd2ccd7fb6c4ae6ac149fd107","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-17 12:51:18.000000000","message":"Patch Set 1: Code-Review+2\n\nMerged to master, good to backport.","accounts_in_message":[],"_revision_number":1},{"id":"3c26b920b1c16a108e33dcb9b63fb124009dd648","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2020-09-17 13:40:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8d7a8335b060edb79e69f4639d6936b50054a165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 14:21:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f3ae5e7bc7484a02abd804929b2a6f68 : SUCCESS in 1h 12m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35de47d3292146859d03c1efa040caa1 : SUCCESS in 1h 11m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/585224776ce54af9a62916e0541146f9 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9632aeff70e442daaa9b1fed2c9ee559 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4c178f732104377a43d302711df6172 : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4062f70de894ff3a8ba9b996adadc4e : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b6456e6c8b40589d2b3b5001f2a1bf : SUCCESS in 8m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b94ddd55901047a781ab6219a7ee66c1 : TIMED_OUT in 31m 47s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/74add1116e4d4e629596cdfad58b8254 : SUCCESS in 9m 35s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/107947129cca4074807283a04b984284 : SUCCESS in 1h 18m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fd292ae0dd945549708533bdc1536e6 : SUCCESS in 1h 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"cb846a8eb6bd9c5242df5aad8ff94cf71182dce3","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2020-09-17 17:59:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5b4c9689c437847e37f156772d73d2cab21ecfe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 20:10:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/129df8685dd64aa3bff25c161024ce2c : SUCCESS in 59m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c10f5976bcf42708c40b3ac4daae124 : SUCCESS in 1h 08m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32734c4b30ab4c43a1ac3d47ed00c262 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8096d73de414fadb52b7f7c92962907 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a91b5d2ca19459c9861600895527a07 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c077261b1e984b89a5f981ab180065d7 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/259c5fd23cf64f5f98880ef137c97280 : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7369937a2b314fdd92569611b3fa1fb0 : SUCCESS in 7m 57s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/93705897168d45c899b033e41f36fcfa : SUCCESS in 11m 32s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9348ba2d1d624805a74beb8cf929b41c : SUCCESS in 1h 00m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6459dfd1d114c4d8e0aab7f001f3081 : SUCCESS in 51m 54s","accounts_in_message":[],"_revision_number":1},{"id":"fe22f7ceb90aa961d40ed9783206086188023622","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-18 06:06:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d9c89bf224e3c8591a6b361fdb024b81a1da6dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 06:07:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c2834b7715bbb00bba383a62cd15f1f6fa5e9c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 07:15:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7193b2b1dee69f58582cdcd3f4699939d330d5f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 07:15:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/df14562859e6475d86b9e3df0f3fca1a : SUCCESS in 1h 03m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cf7307446c2d4f13a4b39e4e54fbe444 : SUCCESS in 59m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7488d9f28ec420cb6094b6f68fefc2d : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa2af071f6a44c9dbbd1c50845f4fbdd : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5be0d4a0d44471bbc3e59066ac4225a : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4dd202dcb03d44c4a40b21d6918a3e63 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e88cdbeb23204499a58a8fefd62043ca : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7786032c3f8e4b65aa7625c1e0c7df0d : SUCCESS in 7m 41s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1a2de164108d493faffea6a80b1b563f : SUCCESS in 7m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae8140a04da543c4937b2d423c5b688c : SUCCESS in 44m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2b3ef1d5dc4a7841940b606eb2174d7c95912049","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 07:17:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc762375959048bca16ac596a3e03a4c : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"977ce4d79cb8e704aa6b5b1d2d983a9509168e8d","revisions":{"977ce4d79cb8e704aa6b5b1d2d983a9509168e8d":{"kind":"REWORK","_number":1,"created":"2020-09-17 08:36:14.000000000","uploader":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"ref":"refs/changes/92/752392/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/752392/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/752392/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/752392/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/752392/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/752392/1"}}},"commit":{"parents":[{"commit":"7ba0762dd80fd35ae97523b15493be76c79e31bb","subject":"Add functional test for task status updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ba0762dd80fd35ae97523b15493be76c79e31bb"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2020-05-02 05:35:12.000000000","tz":540},"committer":{"name":"Margarita Shakhova","email":"shakhova.margarita@gmail.com","date":"2020-09-17 08:36:14.000000000","tz":0},"subject":"Fix a failure to parse json file","message":"Fix a failure to parse json file\n\nChange-Id: I53454c05710ebc45381315a20e10e627cbb65650\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\nCloses-Bug: #1876419\n(cherry picked from commit ed3bb4e1fb07c2817b4b7ad2e3e069f78bd839ba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/977ce4d79cb8e704aa6b5b1d2d983a9509168e8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/977ce4d79cb8e704aa6b5b1d2d983a9509168e8d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
