)]}'
{"id":"openstack%2Fglance~753551","triplet_id":"openstack%2Fglance~master~I765d1ac14c0e1a7c4d67c6b694ec45ea4b4a9d0e","project":"openstack/glance","branch":"master","topic":"bug/1308484","hashtags":[],"change_id":"I765d1ac14c0e1a7c4d67c6b694ec45ea4b4a9d0e","subject":"Member_id validation support for member create command. Bug: bug/1308484","status":"ABANDONED","created":"2020-09-23 09:31:22.000000000","updated":"2022-01-12 15:56:38.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a21f1d245426390c96a516dc0275ae8cae4b98ab","_number":753551,"virtual_id_number":753551,"owner":{"_account_id":31733,"name":"jayaditya gupta","email":"jayadityagupta11@gmail.com","username":"jgupta"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-09-23 09:31:42.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-23 12:21:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-1,"date":"2020-09-23 12:43:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 09:31:42.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-23 12:21:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-23 12:43:28.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"4f05f7f45288a72edbe11a2da01906e0b2f16ed1","author":{"_account_id":31733,"name":"jayaditya gupta","email":"jayadityagupta11@gmail.com","username":"jgupta"},"date":"2020-09-23 09:31:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e1cf6dc63e1ec595ead5384fc0fa1b9a0fc926c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-23 09:31:42.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"39087f0251e6d6f6a961c58a4cb0375f7204120d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 12:21:40.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ee5af8f001ff4062acd05085f9f3a783 : FAILURE in 53m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/41820a5dc2ff409da2882a38befb244b : FAILURE in 1h 04m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cc68692f845457e98c5d24db3995078 : FAILURE in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7199e003b3b0420cb3d8113f004509da : FAILURE in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a91822ed7fea44b78c0222ef1af77b1a : FAILURE in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/851416cb14d54ffea8decb1d8dc215ad : FAILURE in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78cf0a4be3aa44a0b7c4698727bff618 : FAILURE in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/50a208f19ede48ebb4c6e4a005d1fc2c : FAILURE in 28m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ed50a8fd1c584a30a23f6a3100e6e488 : FAILURE in 30m 49s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d89a52f63fb34f659302631b6b898a6f : SUCCESS in 4m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2fd7de2eddad4ebc96c48419f22959b4 : FAILURE in 1h 00m 10s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/fdfea82af2ed43bca709ba9606a34051 : FAILURE in 1h 03m 59s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/e92635716a9a4e03b0c5d281ad8f6553 : FAILURE in 26m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82f4aacb87cd4eec98ea4ced54cec221 : FAILURE in 49m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/40ce1cc4431b4d25823fdf9c8b3c9d37 : FAILURE in 1h 02m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c01672727f5cf710ac5cddbdfc138a060802fd72","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-23 12:43:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nPlease run tox locally before submitting.\n\nLooks like most of this may have been copy/pasted from nova? Make sure to update the code to be correct within glance.","accounts_in_message":[],"_revision_number":1},{"id":"a21f1d245426390c96a516dc0275ae8cae4b98ab","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31733,"name":"jayaditya gupta","email":"jayadityagupta11@gmail.com","username":"jgupta"},"date":"2022-01-12 15:56:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6326cbaf7c8550a2bd86ce20da7cb7b7a95db612","revisions":{"6326cbaf7c8550a2bd86ce20da7cb7b7a95db612":{"kind":"REWORK","_number":1,"created":"2020-09-23 09:31:22.000000000","uploader":{"_account_id":31733,"name":"jayaditya gupta","email":"jayadityagupta11@gmail.com","username":"jgupta"},"ref":"refs/changes/51/753551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/51/753551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/753551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/753551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/753551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/51/753551/1"}}},"commit":{"parents":[{"commit":"922c2ed5adbfedba17324174369cfa2329e999ab","subject":"Fix cleaning of web-download image import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/922c2ed5adbfedba17324174369cfa2329e999ab"}]}],"author":{"name":"jgupta","email":"jayadityagupta11@gmail.com","date":"2020-09-23 09:29:55.000000000","tz":120},"committer":{"name":"jgupta","email":"jayadityagupta11@gmail.com","date":"2020-09-23 09:31:20.000000000","tz":120},"subject":"Member_id validation support for member create command. Bug: bug/1308484","message":"Member_id validation support for member create command.\nBug: bug/1308484\n\nChange-Id: I765d1ac14c0e1a7c4d67c6b694ec45ea4b4a9d0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6326cbaf7c8550a2bd86ce20da7cb7b7a95db612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6326cbaf7c8550a2bd86ce20da7cb7b7a95db612"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
