)]}'
{"id":"openstack%2Fglance~754748","triplet_id":"openstack%2Fglance~master~Ia3fb8aba83bbd7a1399aef136ce8857b14d08435","project":"openstack/glance","branch":"master","topic":"fix-wsgi-defaults","hashtags":[],"change_id":"Ia3fb8aba83bbd7a1399aef136ce8857b14d08435","subject":"Adjust jobs for devstack WSGI mode default","status":"MERGED","created":"2020-09-28 15:40:03.000000000","updated":"2020-09-30 04:53:54.000000000","submitted":"2020-09-30 04:49:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754748-1601441385163-d888da8a","meta_rev_id":"d748710c61d018ba8fffd5b36fa5a1a1a999fd59","_number":754748,"virtual_id_number":754748,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-09-29 20:58:16.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2020-09-29 17:13:44.000000000","_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-30 04:49:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-29 17:08:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-29 16:59:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2020-09-29 17:10:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-29 17:12:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-29 16:59:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 16:59:01.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-09-29 17:10:26.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-09-29 17:12:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-29 20:58:16.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-30 04:49:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d935c17053670a8304b948c65d2d1121ff82603","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-28 15:40:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e303d1eacf5ee72fcbd3bdb21399d0662f11659","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-28 17:22:32.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/48/754748/1/check-not-vote/ext-glance-zuul/3517112 : FAILURE in 8m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7b8c8cf905725c64ed25660d8bfe51423b03172d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 19:13:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8e9df25e03848d8a389bc90dbea56e3 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a90d87c18c348ee9b9ed81f78c9462d : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/905b5bbbd8d44c2ebfe339926e732e47 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/176e7f8f4f244069a8a0aa755133bd05 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/275967c0cbea4f9a950a04425ae0eee4 : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/672518161caa496d80c2c2fcb36836b1 : SUCCESS in 8m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bfc5209e53974c73a8a3f7b3000f229e : SUCCESS in 8m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0bfbac36b2184042a84cf386bac1fc63 : SUCCESS in 1h 02m 20s","accounts_in_message":[],"_revision_number":1},{"id":"9132030564c9f09596e69905310c6e0c08395c1b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-29 13:41:06.000000000","message":"Patch Set 1:\n\nTest patch confirms this is all good: https://review.opendev.org/#/c/754840","accounts_in_message":[],"_revision_number":1},{"id":"5dc3067208e91aba3aa4e4ba4f0349e41058b487","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-29 13:43:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"304dbf17911be35e2a5e70eb109de5e107789b92","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-29 13:44:02.000000000","message":"Patch Set 1:\n\nEr, strike that, I didn\u0027t change the job defs in the check/gate list","accounts_in_message":[],"_revision_number":1},{"id":"a5c2f4b1926b10772b892fb5c0dd83eeba011b64","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-29 14:02:42.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/48/754748/2/check-not-vote/ext-glance-zuul/d76683b : FAILURE in 18m 47s","accounts_in_message":[],"_revision_number":2},{"id":"368b1afcc6a659d0e53e03e48dca4511026ad3a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 14:54:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f4fc03779f14e2683471330aadf8dfb : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca26d0aabf1f4398b3fcb82939f656ee : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c05d67b5f924b17ba5cd25efcd34cf8 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01dd86c44d13487a83cdb974743f7bff : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b4c35c876c4b3986f8841f8b887029 : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2831eca21a624e1d8f217d8d5c318a71 : SUCCESS in 7m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9eae1869f3e343bcb7b1e8402222e0ea : SUCCESS in 8m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/887b55a10b554eaaa37cf801cc7b549d : SUCCESS in 1h 03m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f95aa3705ab64766a8dfeb9f6df31914 : SUCCESS in 1h 01m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4eca74bb0d9d46c8a25c5e85319bcd0d : SUCCESS in 1h 08m 35s","accounts_in_message":[],"_revision_number":2},{"id":"e969a4331cf414fc25cbcd0d95f1c69558981d04","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-29 16:59:01.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\nLooks good, and the jobs are reporting as passing.","accounts_in_message":[],"_revision_number":2},{"id":"05a6564877c84282568df48aeecf2ba0ff7c9f73","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-29 17:08:46.000000000","message":"Patch Set 2: Code-Review+2\n\nThis looks good to me. Waiting to +W until we figure out the devstack situation since that has already branched stable/victoria.","accounts_in_message":[],"_revision_number":2},{"id":"c0251ac41753026e446158546fa999fa6be8ab13","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-29 17:10:26.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)\n\nlgtm, few things for future improvement.","accounts_in_message":[],"_revision_number":2},{"id":"72082b6ef2635916425240baf16e3a8e52273539","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-29 17:12:25.000000000","message":"Patch Set 2: Workflow+1\n\nBoth devstack and glance have a stable/victoria branch, so it is safe to merge this now and then decide whether to backport to stable/victoria.","accounts_in_message":[],"_revision_number":2},{"id":"44b207d23b5d284df0b773644ab158528021208f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-09-29 17:13:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f8834a09932004b6a928ade06c3328c268da0edd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 17:15:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"75a915aae1a30c972b3f3028f87b208a59137f5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 18:49:21.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af5544cf51cd465eba59c4d97586bee5 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39d9d78461854c4abcb28cfebfa99683 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b6a5cd4634a4e838bdc5a50a027f0d9 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c0b72018e8647e2bea311c87f93bedb : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89d7c374fa604ff1b7b59d6ce1212daa : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/35eacb0bc3f548eb99d6e36fc12f1673 : SUCCESS in 8m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3535784382d44881a84027c57b19203b : SUCCESS in 18m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e0c237e9631d42b0bf91c5627705d751 : SUCCESS in 1h 13m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/56308f629fe24e6db02b6303b64f1151 : SUCCESS in 1h 11m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/772aef3958274b95b441d695fe09662e : FAILURE in 1h 06m 18s","accounts_in_message":[],"_revision_number":2},{"id":"82ee1f8677c08488a54d8a215c0dc8fb2c783bf8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-29 20:41:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"34f0a26c4eba475ebe6fc6f92807235b60a12b07","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-29 20:58:16.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/48/754748/2/check-not-vote/ext-glance-zuul/4e7c894 : FAILURE in 16m 29s","accounts_in_message":[],"_revision_number":2},{"id":"e33a1b7a97fcbbc791251a519ef9df9daddb2def","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 22:16:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbb700bcef854d0fba397cf26b3ae04a : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c115a275da3946e092db7a022f9b5c34 : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53941ccc1b5841fda5a61242b21313ff : SUCCESS in 8m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2815f4f78a5241098e01065ff5ee3f53 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae87099ef5d3447b8ae7c494d9b69848 : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45f9728be8f24bd6841729a16867800b : SUCCESS in 12m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/afa3e83a7c9642f5af7be992ede9092d : SUCCESS in 15m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c947be5999f24075ab98905eebc6f51c : SUCCESS in 1h 06m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d2dfdde25af54b12b0c6ebd4d96c2a7d : SUCCESS in 1h 18m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c7d90d2c417748bd8f2c2a86e87419fc : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":2},{"id":"c04cacd3628a704935877c4312539b007c08c01c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 22:17:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5996860b80e0ab55bec8818898f15f9554fc8aea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 23:45:59.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b374efa6a2e4380115f1057ed908d237c0f60056","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 04:49:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c905e5df233a40f0af6e323de7048aa4 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e90412d9559455a9933f77263ce5a82 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0de02ae0ea5944be817f4bd176f57a9f : SUCCESS in 9m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a99f38228d82464895907f9848cea90a : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2df44fbeacb461e937e6a684c6cd772 : SUCCESS in 7m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9ff1cd22976e4770aa7ac840d1eb1802 : SUCCESS in 9m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/27f6ee7d80e0416dbd1cf0c5c7aa2fce : SUCCESS in 8m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3edd06e8cf7f498f9a7a8bff6c756215 : SUCCESS in 1h 09m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fdd5d619ca294c98b5bea53ee04fcbec : SUCCESS in 1h 18m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6c61f8df692e4544804348efb9e54dad : SUCCESS in 1h 07m 49s","accounts_in_message":[],"_revision_number":2},{"id":"e4cd368b0698f2814d280a4b5e7a17d0a0af1cfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 04:49:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4d17d480b0a86638d0c676253ad39ab3c1382d04","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 04:53:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a784b35b7bb4b0a98e66804a6187a09 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b49c0424672ac8191adda5018957a7317fae9de1","revisions":{"2ee8e4644a8c5d823145ead80490aa6fe52512bd":{"kind":"REWORK","_number":1,"created":"2020-09-28 15:40:03.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/48/754748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/754748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/754748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/754748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/754748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/754748/1"}}},"commit":{"parents":[{"commit":"6504588aaa849b89f4b6dfc22d17ee2f8e3b80f2","subject":"Victoria RC-1 release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6504588aaa849b89f4b6dfc22d17ee2f8e3b80f2"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-09-28 15:34:59.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-09-28 15:39:21.000000000","tz":-420},"subject":"Adjust jobs for devstack WSGI mode default","message":"Adjust jobs for devstack WSGI mode default\n\nDevstack is switching the default glance operational model back to pure-WSGI\nmode for parity with the rest of the services. This adjusts our jobs to fit\nwith that so that we\u0027re testing the default, but also the standalone case.\nAfter this change, we end up with the following for the vanilla devstack\njobs:\n\n - The \"-import\" job, which takes the devstack default of WSGI mode,\n   with all the tweaks required for import to be enabled. It configures\n   import conversion to raw, metadata injection, and confirms that with a\n   post-run check for that metadata.\n - The \"-import-standalone\" job, which inherits all the behaviors from\n   the job above, but configures glance in standalone mode.\n\nDepends-On: https://review.opendev.org/#/c/742884\nChange-Id: Ia3fb8aba83bbd7a1399aef136ce8857b14d08435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2ee8e4644a8c5d823145ead80490aa6fe52512bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2ee8e4644a8c5d823145ead80490aa6fe52512bd"}]},"branch":"refs/heads/master"},"b49c0424672ac8191adda5018957a7317fae9de1":{"kind":"REWORK","_number":2,"created":"2020-09-29 13:43:35.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/48/754748/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/754748/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/754748/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/754748/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/754748/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/754748/2"}}},"commit":{"parents":[{"commit":"6504588aaa849b89f4b6dfc22d17ee2f8e3b80f2","subject":"Victoria RC-1 release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6504588aaa849b89f4b6dfc22d17ee2f8e3b80f2"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-09-28 15:34:59.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-09-29 13:43:27.000000000","tz":-420},"subject":"Adjust jobs for devstack WSGI mode default","message":"Adjust jobs for devstack WSGI mode default\n\nDevstack is switching the default glance operational model back to pure-WSGI\nmode for parity with the rest of the services. This adjusts our jobs to fit\nwith that so that we\u0027re testing the default, but also the standalone case.\nAfter this change, we end up with the following for the vanilla devstack\njobs:\n\n - The \"-import\" job, which takes the devstack default of WSGI mode,\n   with all the tweaks required for import to be enabled. It configures\n   import conversion to raw, metadata injection, and confirms that with a\n   post-run check for that metadata.\n - The \"-import-standalone\" job, which inherits all the behaviors from\n   the job above, but configures glance in standalone mode.\n\nDepends-On: https://review.opendev.org/#/c/742884\nChange-Id: Ia3fb8aba83bbd7a1399aef136ce8857b14d08435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b49c0424672ac8191adda5018957a7317fae9de1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b49c0424672ac8191adda5018957a7317fae9de1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
