)]}'
{"id":"openstack%2Fglance~755088","triplet_id":"openstack%2Fglance~master~I1e200c47f20ec6945981dbd92f859cfec234e42d","project":"openstack/glance","branch":"master","topic":"api_version_master","hashtags":[],"change_id":"I1e200c47f20ec6945981dbd92f859cfec234e42d","subject":"Bump Images API version to 2.11","status":"MERGED","created":"2020-09-29 17:07:38.000000000","updated":"2020-10-02 06:31:31.000000000","submitted":"2020-10-02 06:29:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755088-1601620193740-e452fb22","meta_rev_id":"4bd2a7712518ae489c52ce556b8aae2a8ab4f1a4","_number":755088,"virtual_id_number":755088,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-10-02 04:07:42.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-02 06:29:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-01 13:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-10-01 15:16:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-29 22:03:37.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-09-30 15:10:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-29 22:03:37.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-10-01 13:07:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-02 04:07:42.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-02 06:29:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6618ce9c65be2a21b186d6e66afa84046065092b","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-09-29 17:07:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb926cde9aea3e9cce4eb1366d52524803a44f4d","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-09-29 17:08:02.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b594633a9002671ab610264f251a0f1d82944168","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-29 17:29:52.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/88/755088/2/check-not-vote/ext-glance-zuul/7019a66 : FAILURE in 21m 31s","accounts_in_message":[],"_revision_number":2},{"id":"e25c7bae1e20ceea3373a2d676d2eb18acfc8338","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 19:48:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/af3b6d1f217a42dba4cdfb9d97ed7de9 : SUCCESS in 57m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c4d3be91d96841d78279ecdf40b4c373 : SUCCESS in 1h 10m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab1cf21b494246df8eedadf32cf07c9f : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2226e5aef9847f0921f42f4b640ab11 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4f86b9de4414357bee9d0285fc803bb : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c43da76b154248bca9b212646303c78b : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c70323dd42d458dbf5be7c9f4d86a99 : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/77cc65a54371449592e09a903e4aa2dd : SUCCESS in 7m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bd0e43f0cc7743f8b5ec95b7a72d6f8d : SUCCESS in 7m 40s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6f9162bedd7c4810b42dbb10686a337e : SUCCESS in 3m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/65b35a8ddba049feadb8cb6ce610a080 : SUCCESS in 1h 04m 33s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/dc5bddda91114bfba58d7cb9418f4c51 : FAILURE in 1h 25m 35s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/b80fa55e50454224beeab9d5c46653e5 : SUCCESS in 1h 05m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/510be9477ae443a095fa6532987e6995 : SUCCESS in 59m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f4214afdeeff4767866a8cc9a4abd4a3 : SUCCESS in 1h 08m 22s","accounts_in_message":[],"_revision_number":2},{"id":"423fd392d712a70493f57b323f25b2d1b82aa381","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-29 21:49:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4b2a2e03e5470e4cfb24710eac2d5a8c4bd3cfe4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-29 21:52:35.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nSee comment inline.","accounts_in_message":[],"_revision_number":3},{"id":"8c836fcea5a95e8f4ca9849b6b44cfb7e7282b8d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-09-29 22:03:37.000000000","message":"Patch Set 3:\n\nYou may be wondering why none of the tests failed on the original patch.  Two reasons: (1) they were testing the versions response for the default configuration only, and (2) the v2.10-doesn\u0027t-exist test was passing because the float-to-string conversion problem noted inline really was excluding 2.10 from the versions response.  I think the revisions on this patch will break next time if the versions response is changed without the version_negotiation not also being changed, but I\u0027m not 100% sure, so that\u0027s something to keep in mind when reviewing.","accounts_in_message":[],"_revision_number":3},{"id":"49ffc8f654e07866fcf4794123f8e0fe87be3125","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-29 22:47:57.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/88/755088/3/check-not-vote/ext-glance-zuul/a79902a : FAILURE in 58m 18s","accounts_in_message":[],"_revision_number":3},{"id":"320f9637c21b7870e318f1d68d8266f3836cf4f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 02:09:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/df38a599c21841349ee891f0b701b5b1 : SUCCESS in 1h 03m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8924555b381c4335990bab0610610197 : SUCCESS in 1h 02m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/854322bcf01f469aa6aa6a7d480bb45f : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59946d9c511d40529f9a26ebf0192fc7 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a92c50b8937546ea8924af12454a2123 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8820d2d031b94865b008cdb9202f157b : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd04edfbc4604fb8b1083de5618765cb : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/79c72be6346b40bf908cddafc5c31fc3 : SUCCESS in 6m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e59d14fb74f840a695c3368aa01d7cb5 : SUCCESS in 8m 46s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/19aae80d494c4847aed7d29599a5b691 : SUCCESS in 6m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d51aac7d5a374312a4d62e702dd56d83 : SUCCESS in 1h 06m 28s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/fe3cf0e45b3b4a82928c41a10ec5e0fe : SUCCESS in 1h 05m 27s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/007d6e95eea944b998386f343f6be50a : TIMED_OUT in 2h 12m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be64629c8a354e2f9fec5b0e619c18fc : SUCCESS in 57m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b5f98d4b75e744e1a29b830cb84a7c41 : SUCCESS in 1h 54m 58s","accounts_in_message":[],"_revision_number":3},{"id":"0db40dfc22542eda98caa82a27135553f1684c90","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-09-30 12:01:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"214d1530a91ac8420132826ea234db1707778257","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-30 12:16:49.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/88/755088/3/check-not-vote/ext-glance-zuul/219bd81 : FAILURE in 15m 09s","accounts_in_message":[],"_revision_number":3},{"id":"7b747354fd87bb6a51f357841c52ee4d4e722e65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 14:46:00.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/04406d84b38b4a64ba7696856e304a15 : SUCCESS in 1h 02m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/80448af027fb4d25bf5a3a7d4224fa7d : SUCCESS in 1h 55m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5388d8abcabe4978b70c9e33d4d774d6 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c2e61f4038946e38ba6d666c2efc39f : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8de1791c88042f8807489847bd7595f : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed25bfb7cd3f4b0c9eb1671d28a2ac2c : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ada91bf800c4229a8fa27b40de79430 : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f326a4c1a82c4e6f8770b94cb5c3b714 : SUCCESS in 8m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e641e81549424c8e8a0f842824b278e0 : SUCCESS in 7m 10s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/29fa02f698d14c8d857c900515aec3cc : SUCCESS in 6m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/38832b00a68049d1b3ec8203d5b146bb : SUCCESS in 1h 12m 34s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6fc47916c9d548dca7a672b454151835 : SUCCESS in 1h 11m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5a0af1d0105145f58878f56a8c5fa937 : FAILURE in 59m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36cbee26188347619b03037cbcbb443f : SUCCESS in 1h 00m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e3dddffeda4b4d3eabc180189bb7ecae : SUCCESS in 1h 23m 53s","accounts_in_message":[],"_revision_number":3},{"id":"e584659ab77ed0626de50b0b7b2717d14ff59c56","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-09-30 15:09:25.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"936e27a13a06ed0d0c6b8bd180e8e75f34ad322a","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-09-30 15:10:34.000000000","message":"Patch Set 3: Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"be0660100d7b8bccb7654ced3332c6075e70ebca","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-30 15:41:40.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/88/755088/3/check-not-vote/ext-glance-zuul/6f93c00 : FAILURE in 31m 54s","accounts_in_message":[],"_revision_number":3},{"id":"0910df58bcafd1b9622aadf2af8e15e417826ce4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 21:00:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/6ff07a1bba0a44c0b067e0109d04e314 : SUCCESS in 1h 04m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56a6ed99f2984d299548a5c8e5d2ae9c : SUCCESS in 1h 19m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2bee54b089cc4b799c56ca7456624a5c : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33a66b7c8ae843198e2f7a92f440c0a2 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/813d5493e74740079373d2a54381aa52 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7136f013c5824ec6901a30125e56d828 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0800f26cd784f07bb76f57792093ff1 : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d40a2aa8047f4f1f9272eed438bd3de7 : SUCCESS in 7m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f1d5c2be39394e2e8734470c958f0af3 : SUCCESS in 8m 26s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/aedea6154c5d4a5781ad21a8e10d61ac : SUCCESS in 4m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/60f558acee5d46ec845bcf29fa3ae2e9 : SUCCESS in 1h 05m 23s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/046aaaa3db884b62a3d0bd7864c842f1 : SUCCESS in 1h 05m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/596dcc80d3114968bf79fa2f357fca7d : SUCCESS in 1h 05m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/759c0abb65974be9a9fdf57d9c30ae96 : SUCCESS in 56m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9c5829afc8fd48d9a15996dc5b44db15 : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":3},{"id":"3cfb98c3d939ca1450848d5d746968663d201419","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-01 13:07:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"73c21f657fc59c80c38ea83b55dff6385c82b730","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-10-01 15:16:05.000000000","message":"Patch Set 3: Workflow+1\n\nMostly Brian\u0027s work at this point spotting and addressing the issues the original string conversion had. LGTM","accounts_in_message":[],"_revision_number":3},{"id":"684ec8ffa7cb3f54e3e8b8bb8ee4a1827a97792e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 15:16:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2a7d8afdc7d769c2ed1337d8e47a21b1245ae25e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 22:09:19.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2ddec4640b4142ee85eaf36d9b85d337 : SUCCESS in 1h 04m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6a939bae5d0e4bbca2cf3e17cbcdf736 : SUCCESS in 1h 11m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8293f2dd33d444387550db0010b4191 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb0c7c19e4134d52892ee270c1d3ca54 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0e6c7c1b10f43f7a7867daa8c787ca0 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26799de71df4440eb265e8db776ff597 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d80c22f30c18495aa246d8eca88da11d : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/072ab991f6b14183babbdf4f7129d528 : SUCCESS in 6m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ab038298f9b47fca517a6145b105996 : SUCCESS in 7m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5afd90bed57c4c9d9f33989d5f58d5cb : FAILURE in 54m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/55c59aee4e0a4a4693efb547007efa75 : SUCCESS in 1h 14m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d7d52ba4ca64b7293a6bf4313ddd137 : SUCCESS in 1h 13m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b24cb620f5604a4482a46d71bb1fbd49 : SUCCESS in 1h 05m 07s","accounts_in_message":[],"_revision_number":3},{"id":"d8c2d8763ddbe55a55c93286ebeac193c7a518fb","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-10-01 22:13:50.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9cfee3ad94c1ea1175c992ebc7191acda36b09fe","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-01 22:30:54.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/88/755088/3/check-not-vote/ext-glance-zuul/fd3eebb : FAILURE in 16m 43s","accounts_in_message":[],"_revision_number":3},{"id":"57e1dfce567a9d5d0d569077070368430bdd88f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 01:40:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c465ea6dfad748d69cee4b2ddf96af9b : SUCCESS in 1h 03m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/28caefecf37d44909737402ee98fb0ea : SUCCESS in 1h 09m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85e0e5035afd430a900a0daa8f7327ac : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d28b08ffc4a45cc9deef1c34c308296 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06b4cf15cd8e45498acdf5e1ab51e924 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81259ba2aff548719b50023fa9467996 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c734ec0a5a4e417593fcae8340bc10fd : SUCCESS in 5m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eefe219683a34dfa8b87556d90eee18b : SUCCESS in 6m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c58aafcf5dd64770a0bf3b6c37d6cc79 : SUCCESS in 7m 56s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/32ab3dfa609d48db9f5339cf1abd6e66 : SUCCESS in 4m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3253081d46124753ad27109191752d4e : SUCCESS in 1h 04m 07s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d5de746a55a44256b33ae28c22337cff : SUCCESS in 1h 15m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a1d1ea7ba34e404e800409a7e03afeeb : SUCCESS in 1h 13m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f70d0404572b41d496d7fda92e21485a : SUCCESS in 1h 01m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1ae127a50adb4d869a91a4e95e9af979 : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":3},{"id":"0f084ac866cab0a92b183db816e934f8002d87e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 01:40:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fe39356730bcd715cdad0ac88fb6a385db2484b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 03:10:14.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/387b8e89815b40e2a1d50c8bf882763c : SUCCESS in 54m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/16cae61486dd4ac6a09391a7edd59644 : POST_FAILURE in 1h 05m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b22c7ffdc21c4ddf95db966a076938ee : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92de31a159874f2593a4b5353dd63503 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80fef7aa9a314f4eb0959a549b102abc : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a58ab58d93454f30a0c1fa00bad679aa : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe83e1bf94a34fe08a73785ab6e266b4 : SUCCESS in 5m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1aecff732ba44ed9951d0ac88975c82f : SUCCESS in 8m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a30251d5b5224a7eaf0fb295af866f9e : TIMED_OUT in 30m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/355ec2e95bf14edabf31ec3f0f320823 : SUCCESS in 1h 10m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e21c15e02b2e4e138208a00c01d50fc3 : FAILURE in 1h 28m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55f937974a6f46ef8c935966cb3d963e : SUCCESS in 1h 02m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5e68ced69bce4cc6b41bb6c9d6b5b59e : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":3},{"id":"7faea7415ab1ad10fc146524a3520cfc95ec24dd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-02 03:32:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5384a425aef5506c511adc0b065fc05cf480408c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-02 03:47:44.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/88/755088/3/check-not-vote/ext-glance-zuul/6052ab6 : FAILURE in 15m 17s","accounts_in_message":[],"_revision_number":3},{"id":"a779172399fb3d35d21def900b31f4b30797ad41","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-10-02 03:51:08.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cf95668e2bfa1e4a67cd3372d2433a8f52f62860","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-02 04:07:42.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/88/755088/3/check-not-vote/ext-glance-zuul/e70cf89 : FAILURE in 16m 14s","accounts_in_message":[],"_revision_number":3},{"id":"6492f6cd893901b1fe9ce9b78e4ffc966788df67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 05:06:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/67759dc2755e4795b68d53a6f1eb0e19 : SUCCESS in 1h 32m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56869ac3a73140b58a285c86728e826b : SUCCESS in 1h 10m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e15124f1d4154ca0bd4bebe5563dd0b5 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e69733a8e0fc4e7db88a82f1e85d06dd : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/573447214ad14253b2ebe5349817cec4 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24bfb76761044f7fb6783fcb21a9c845 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eaecb834b7a4217a9fce7596a0e7518 : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6611bc97d5714024aa538d6cab36a299 : SUCCESS in 7m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9365865d04c6492eb22e3594cf568eb0 : SUCCESS in 7m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8df6dfd678634cdbbf1e7958dba0aa87 : SUCCESS in 3m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/816b1b7d8e044be79f9258dd7e51f819 : SUCCESS in 1h 11m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c1b7cab647da4a9987cf83464a2c9177 : SUCCESS in 1h 07m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fcb990cbd4b3435e81e87561a11b62ae : SUCCESS in 1h 14m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/183e5d39b4884165914f6c8599dd8bd0 : SUCCESS in 1h 00m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3433d77b349d4b35ae7dd17e966addcd : SUCCESS in 1h 11m 25s","accounts_in_message":[],"_revision_number":3},{"id":"42e89f9889ec84f33b035ff10fa6126513ac5f8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 05:06:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f2513feefb4b158a5a6f2db382b44b6ff8cd0940","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 06:29:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b741114ac2714873b3452dbfaf4119de : SUCCESS in 1h 08m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb4a2f1650aa458099f2606cb49242c4 : SUCCESS in 1h 16m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae1ba0c73d104a00a8274a6c9955d7d3 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/202ea3ad42734cc7a22c5052fa7ad408 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a51ca561fce481abebe50cee46431d8 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b750e2c3aff240b294e004994ab5d210 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/435057f1fba0478e9d7636602f2ce63a : SUCCESS in 8m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e27962a8b9eb44f5acff5c2931d16637 : SUCCESS in 7m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/549f5688e1b24b4695e41c390ecb6f7c : SUCCESS in 10m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ce10f545a16b4fdfb873992e09ab690e : SUCCESS in 1h 07m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/345ca45681e245c781b138390ceed2f1 : SUCCESS in 1h 13m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e7bff686804407680a9e074b0c5fdcb : SUCCESS in 1h 05m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b78bea7199f04fe597c15c661c1f6e57 : SUCCESS in 1h 11m 01s","accounts_in_message":[],"_revision_number":3},{"id":"4da6dccc82364fd5812079e28c7d97f7221101af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 06:29:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0d2ad850584076fa56e6f7d9548827eadbcd2490","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 06:31:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43bc3302742546bfbf736e42ec4c3159 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0e3d274915699d0d38c45fa88a1f0c76a3254ab5","revisions":{"d7e4f906026d876824200813a9c559dc6bde1338":{"kind":"REWORK","_number":1,"created":"2020-09-29 17:07:38.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/88/755088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/755088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/755088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/755088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/755088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/755088/1"}}},"commit":{"parents":[{"commit":"6504588aaa849b89f4b6dfc22d17ee2f8e3b80f2","subject":"Victoria RC-1 release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6504588aaa849b89f4b6dfc22d17ee2f8e3b80f2"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-09-29 17:04:59.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-09-29 17:07:10.000000000","tz":60},"subject":"Bump Images API version to 2.11","message":"Bump Images API version to 2.11\n\nChange-Id: I1e200c47f20ec6945981dbd92f859cfec234e42d\nCloses-bug: #1897773\n(cherry picked from commit a51301b69f7d500e5ad4e82bcdc3fa57b86c54c4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d7e4f906026d876824200813a9c559dc6bde1338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d7e4f906026d876824200813a9c559dc6bde1338"}]},"branch":"refs/heads/master"},"a51301b69f7d500e5ad4e82bcdc3fa57b86c54c4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-29 17:08:02.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/88/755088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/755088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/755088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/755088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/755088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/755088/2"}}},"commit":{"parents":[{"commit":"6504588aaa849b89f4b6dfc22d17ee2f8e3b80f2","subject":"Victoria RC-1 release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6504588aaa849b89f4b6dfc22d17ee2f8e3b80f2"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-09-29 17:04:59.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-09-29 17:04:59.000000000","tz":60},"subject":"Bump Images API version to 2.11","message":"Bump Images API version to 2.11\n\nChange-Id: I1e200c47f20ec6945981dbd92f859cfec234e42d\nrelated-bug: #1897773\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a51301b69f7d500e5ad4e82bcdc3fa57b86c54c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a51301b69f7d500e5ad4e82bcdc3fa57b86c54c4"}]},"branch":"refs/heads/master"},"0e3d274915699d0d38c45fa88a1f0c76a3254ab5":{"kind":"REWORK","_number":3,"created":"2020-09-29 21:49:18.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/88/755088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/755088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/755088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/755088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/755088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/755088/3"}}},"commit":{"parents":[{"commit":"6504588aaa849b89f4b6dfc22d17ee2f8e3b80f2","subject":"Victoria RC-1 release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6504588aaa849b89f4b6dfc22d17ee2f8e3b80f2"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-09-29 17:04:59.000000000","tz":60},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-09-29 21:47:45.000000000","tz":-240},"subject":"Bump Images API version to 2.11","message":"Bump Images API version to 2.11\n\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I1e200c47f20ec6945981dbd92f859cfec234e42d\nrelated-bug: #1897773\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e3d274915699d0d38c45fa88a1f0c76a3254ab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e3d274915699d0d38c45fa88a1f0c76a3254ab5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
