)]}'
{"id":"openstack%2Fglance~755090","triplet_id":"openstack%2Fglance~stable%2Fvictoria~I4e23ef8c7eba92addc192593f38fbc5f7ba7b03e","project":"openstack/glance","branch":"stable/victoria","topic":"api_version_victoria","hashtags":[],"change_id":"I4e23ef8c7eba92addc192593f38fbc5f7ba7b03e","subject":"Bump Images API version to 2.11","status":"MERGED","created":"2020-09-29 17:14:31.000000000","updated":"2020-10-01 23:07:55.000000000","submitted":"2020-10-01 23:06:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755090-1601593595817-5cec80e2","meta_rev_id":"77a86e0afd57aa9368d756cd3fc964915864b128","_number":755090,"virtual_id_number":755090,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-01 23:06:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-01 13:11:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-10-01 15:18:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-09-30 22:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 15:18:26.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-01 23:06:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d555142d9c7dfa74cb9362848a9009def0d3efce","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-09-29 17:14:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ccf68523ce3a64880e17f00ca13a876015deae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 19:42:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7fc06e09952455f8ad4a92adff2a066 : SUCCESS in 1h 03m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b214448166c8492c83075451eebb1603 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c053f339b8fd4ea99fe415283133def4 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/caf9cd42c584431ca9dd62eec2a8f3ff : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/006a8fb7c4944fc697b72c618b49b1d2 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b37de506b6a04049ad5e941c85c1837f : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fca2e03f8c8d4ae282f6f5fecca315dc : SUCCESS in 7m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9e61e405bbf841c8977795e90bdb6012 : SUCCESS in 7m 11s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0441323a4a0e421780dda6137f1bec88 : SUCCESS in 5m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/454dad22ac24408084300628806c4f61 : SUCCESS in 1h 03m 35s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/daa41829aaad48d5a6986a26d1bc8c79 : SUCCESS in 1h 14m 13s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/5d0713d26d264fa8adde61f6e8484cfd : SUCCESS in 1h 08m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be97ef57c4b44119958a6e074b8213da : SUCCESS in 1h 08m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/31ebbe67973c4cc4a3bfd152ef4c1d1c : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":1},{"id":"76081fedc553331aa9175ca6f9da2d3dcb71fbc9","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-09-30 15:10:49.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"3385f2eadddbce085fbc7498233b844121f9c703","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-09-30 15:10:58.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"264835790c8d65b6b8a5b2321f4bb607a1fd176b","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-09-30 22:46:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ae92d7a05250811bae42e178c2afd991def2d12e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 00:50:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/340a8646d112421aba97d668ca478b4a : SUCCESS in 1h 12m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b79940de5cf244ed93247d3ef0e45cf8 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7468d5a594c41b5b08a3964e091034c : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e25bac0db1c244299be269135ad70e1a : SUCCESS in 10m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b057213c3c84d3a9501c9f010dfa87c : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd2b86a1b67047fa9e6b486c682294a7 : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/df518fe6efef4bc4b17c68c330860e6c : SUCCESS in 7m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/779010cd67a1452b958a14388449940c : SUCCESS in 7m 42s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d4a919f0414948bd8dac6d45c4342b03 : SUCCESS in 4m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/00632d01240d42bcafa3e55b0c6e9493 : SUCCESS in 1h 02m 02s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/975a839e00e54ffba7272599482a4c32 : SUCCESS in 1h 23m 35s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/c24b932cffa1414aa7a25f289b425616 : SUCCESS in 1h 22m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0f3c3f064a6443a92613848edb0efcc : SUCCESS in 57m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d1d6dc2a67764831b81323cd4438e951 : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":2},{"id":"033c3eee1d668e5098ff8b06dcd9981d812817d4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-01 13:11:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"904c29f5ff02dfeea5bc909b5135b50e6ceca943","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-01 15:18:26.000000000","message":"Patch Set 2: Workflow+1\n\nFolks are not around right now, and this is important to unblock some things, so single approving. Feel free to revert if I\u0027ve made a mistake, but this is now merging to master, so should be good to go in stable/victoria.","accounts_in_message":[],"_revision_number":2},{"id":"b271b812064b957084d8d8bb0fac6bbdc12910f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 15:18:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87b3d29cd7d8e4259dc55f8c6e41fba91c21c8c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 23:06:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"21d84f55edf7c920fc517b9b1be34f6c59606c2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 23:06:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/29ac3e3db44745c58e0439f1a660e895 : SUCCESS in 1h 01m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/06c72578d8704db1ad133531e05a12cc : SUCCESS in 1h 09m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9616b88820da406da7e91014f8d04f81 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c675880957a94cebb41c81d8df2b600f : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a15df19b95d44999b2122bf59c534ed9 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c625dfcff888427aa01bae482a4bd5de : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05e164d19920498ebdb751c5a1c20129 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0ba4d68f85144b57b795bb4514eb3702 : SUCCESS in 8m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4e0a6b84b7b04add8ed953444bce85ba : SUCCESS in 8m 28s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/7cda4dcc9e464885b78feaf11869f508 : SUCCESS in 1h 09m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afe16ac8514341deb2729b62f8732203 : SUCCESS in 1h 06m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2c8de6d9664b444b98ea47fc0b659938 : SUCCESS in 1h 20m 32s","accounts_in_message":[],"_revision_number":2},{"id":"d7b1ae3ada777b4288908eaeabf33145e3a133a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 23:07:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef726c5ba4094a7bb59ebf81fb641086 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c4d8880794c20aad6772f6e9c2217f15f41c9c2c","revisions":{"8de0c830edcb9c3e7b17d12ca0d9e478743cea26":{"kind":"REWORK","_number":1,"created":"2020-09-29 17:14:31.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/90/755090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/755090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/755090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/755090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/755090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/755090/1"}}},"commit":{"parents":[{"commit":"6504588aaa849b89f4b6dfc22d17ee2f8e3b80f2","subject":"Victoria RC-1 release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6504588aaa849b89f4b6dfc22d17ee2f8e3b80f2"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-09-29 17:04:59.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-09-29 17:14:10.000000000","tz":60},"subject":"Bump Images API version to 2.11","message":"Bump Images API version to 2.11\n\nCloses-bug: #1897773\n(cherry picked from commit a51301b69f7d500e5ad4e82bcdc3fa57b86c54c4)\n\nChange-Id: I4e23ef8c7eba92addc192593f38fbc5f7ba7b03e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8de0c830edcb9c3e7b17d12ca0d9e478743cea26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8de0c830edcb9c3e7b17d12ca0d9e478743cea26"}]},"branch":"refs/heads/stable/victoria"},"c4d8880794c20aad6772f6e9c2217f15f41c9c2c":{"kind":"REWORK","_number":2,"created":"2020-09-30 22:46:55.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/90/755090/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/90/755090/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/755090/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/755090/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/90/755090/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/90/755090/2"}}},"commit":{"parents":[{"commit":"6504588aaa849b89f4b6dfc22d17ee2f8e3b80f2","subject":"Victoria RC-1 release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6504588aaa849b89f4b6dfc22d17ee2f8e3b80f2"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-09-29 17:04:59.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-09-30 22:45:36.000000000","tz":60},"subject":"Bump Images API version to 2.11","message":"Bump Images API version to 2.11\n\nCo-authored-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n\nChange-Id: I4e23ef8c7eba92addc192593f38fbc5f7ba7b03e\nrelated-bug: #1897773\n(cherry picked from commit cb2fb167f4ab17b2168177f47c4325afbc3d9765)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c4d8880794c20aad6772f6e9c2217f15f41c9c2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c4d8880794c20aad6772f6e9c2217f15f41c9c2c"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
