)]}'
{"id":"openstack%2Fglance~755215","triplet_id":"openstack%2Fglance~master~I8b0869f178c535b58aaf72c63dccab5c50da9f4b","project":"openstack/glance","branch":"master","topic":"glance_bug","hashtags":[],"change_id":"I8b0869f178c535b58aaf72c63dccab5c50da9f4b","subject":"new file added","status":"ABANDONED","created":"2020-09-30 07:56:14.000000000","updated":"2020-09-30 11:34:44.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"845ffa24167554ffb3bb5c416062357cc8cee88b","_number":755215,"virtual_id_number":755215,"owner":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-09-30 08:12:02.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-30 10:11:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-2,"date":"2020-09-30 11:34:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 08:12:02.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-30 10:11:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-30 11:34:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"6aa5ade106f7cda289108ff01487ed90acdc41a3","author":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"date":"2020-09-30 07:56:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa251ab94e6bf95ae7cd0b3eeef8f45ab63ca458","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-30 08:12:02.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/15/755215/1/check-not-vote/ext-glance-zuul/3238d0a : FAILURE in 15m 27s","accounts_in_message":[],"_revision_number":1},{"id":"0d4b847bb1e4a712cc778dcc412cfa5c428495e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 10:11:20.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e0b65a985d74eeba7f60235846e9445 : FAILURE in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a8dd0be3b464230bf4bb34bab6dafaf : FAILURE in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"3a5f1af77251350b3173029251e4d39e22004e5e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-30 11:34:41.000000000","message":"Patch Set 1: Code-Review-2\n\n(2 comments)\n\nAs discussed, update the existing patch instead of creating new ones every time. Run tox locally to test your changes. Address all comments and issues before pushing up updates.","accounts_in_message":[],"_revision_number":1},{"id":"70c6a92584615ea678715434f684591cea61d78e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-30 11:34:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a82b03f5085989f4dfa2c231b65518ef0a43e2f","revisions":{"2a82b03f5085989f4dfa2c231b65518ef0a43e2f":{"kind":"REWORK","_number":1,"created":"2020-09-30 07:56:14.000000000","uploader":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"ref":"refs/changes/15/755215/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/755215/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/755215/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/755215/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/755215/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/755215/1"}}},"commit":{"parents":[{"commit":"b49c0424672ac8191adda5018957a7317fae9de1","subject":"Adjust jobs for devstack WSGI mode default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b49c0424672ac8191adda5018957a7317fae9de1"}]}],"author":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-09-30 07:51:03.000000000","tz":330},"committer":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-09-30 07:51:03.000000000","tz":330},"subject":"new file added","message":"new file added\n\ntitle:need multihash info in glance glance user, admin guides\n\ndescription:new file os_hash_algo is added in user guide and admin guide\n\nChange-Id: I8b0869f178c535b58aaf72c63dccab5c50da9f4b\nCloses-bug:#1871419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a82b03f5085989f4dfa2c231b65518ef0a43e2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a82b03f5085989f4dfa2c231b65518ef0a43e2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
