)]}'
{"id":"openstack%2Fglance~755338","triplet_id":"openstack%2Fglance~master~Id78be3935998b9c5acdd0706393117e892e5ab59","project":"openstack/glance","branch":"master","topic":"glance_bug","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2023-02-09 15:50:48.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-02-09 15:50:48.000000000","reason":"Change was submitted"},"32410":{"account":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"last_update":"2023-02-09 15:50:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id78be3935998b9c5acdd0706393117e892e5ab59","subject":"Add multihash info in glance documentation","status":"MERGED","created":"2020-09-30 16:12:20.000000000","updated":"2023-02-09 15:52:58.000000000","submitted":"2023-02-09 15:50:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755338-glance_bug","meta_rev_id":"4e7a7ef739f9516d8a315a92671775a03316db97","_number":755338,"virtual_id_number":755338,"owner":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-09 15:50:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-02-09 14:35:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-02-09 14:35:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-06 13:37:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-06 13:42:36.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-10-08 14:29:07.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-09 00:32:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 14:28:19.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-03-29 14:28:19.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-29 14:31:30.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-02-08 20:32:37.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"}],"messages":[{"id":"c96cef5dcb6883ca83344e77733131288b85ca32","author":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"date":"2020-09-30 16:12:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b166e74e3036e6f0d86756b21f7f92f1c6fa39f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-30 16:23:36.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/38/755338/1/check-not-vote/ext-glance-zuul/83ce56f : FAILURE in 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"aa7b08355f1753f37fe80e5ebf44621659f37fda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 20:37:05.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e247fb94a1348e3ac397a57e7d871da : FAILURE in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4973c264b4b8466c83ee050d4167a054 : FAILURE in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ba5022fdd0b5e63b13930cda58f7d2aeb155cf34","author":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"date":"2020-10-01 07:48:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ecd7138dcc7b0199ce696b45e3e77be770f746b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-01 08:04:26.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/38/755338/2/check-not-vote/ext-glance-zuul/3014710 : FAILURE in 15m 56s","accounts_in_message":[],"_revision_number":2},{"id":"ed7c3a63f5bf3d0267f98d1fdbb703b06d42d6bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 08:24:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4008fb6d355849c380f79b1bb60d2aa6 : FAILURE in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b17843db22ad4e2ca3c47c6531ebb052 : FAILURE in 5m 57s","accounts_in_message":[],"_revision_number":2},{"id":"1108a1b6641a7c130623f604d51b284fe9d2da26","author":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"date":"2020-10-01 09:30:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"93760b12a33dbfdb9ec602efcd48e34738d3c2dd","author":{"_account_id":30905,"name":"Harshavardhan Metla","email":"harshavardhan.metla@tcs.com","username":"harsha24"},"date":"2020-10-01 09:37:55.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"0fc0da52e839681708accc274f983ae6bcaa2331","author":{"_account_id":30905,"name":"Harshavardhan Metla","email":"harshavardhan.metla@tcs.com","username":"harsha24"},"date":"2020-10-01 09:38:37.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"7cb7a4476eb69b7434b03fbd81169cb5db2cea25","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-01 09:55:48.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/38/755338/5/check-not-vote/ext-glance-zuul/6e020d8 : FAILURE in 16m 46s","accounts_in_message":[],"_revision_number":5},{"id":"3af06a91e273a7c43ec04ae0f96d171a9e64195c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 10:42:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eafbc75471d74c49a258c0438c2a7261 : FAILURE in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6eafbb737b145cbaf7cf9abe0a09b93 : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":5},{"id":"f6efbdf0effe3e9000f36d26bd5b19c654b744b3","author":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"date":"2020-10-01 12:31:23.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"6f140aff3f5297e3b8633745442de4d64c2b880c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-01 12:50:21.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/38/755338/6/check-not-vote/ext-glance-zuul/fdbe26a : FAILURE in 17m 44s","accounts_in_message":[],"_revision_number":6},{"id":"4158dbafebed424f47f5027d8a5780ed9e4b3c00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 13:31:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44d522297c574adc93a70620b367d4e6 : FAILURE in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51e6dbad14ed4e8dab6240db5b8c0db5 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":6},{"id":"4217b1664e3b63c8960c3770e4af3047fdfd4690","author":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"date":"2020-10-01 15:17:12.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"dcde2f791c047b0e784c20e06a5a5cff02719e35","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-01 15:33:03.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/38/755338/7/check-not-vote/ext-glance-zuul/e47cec7 : FAILURE in 15m 26s","accounts_in_message":[],"_revision_number":7},{"id":"76dbcbe5f49f758783abe8ca5fb41f671b354847","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-01 18:40:35.000000000","message":"Patch Set 7: Code-Review-2\n\nThis patch is based on top of an abandoned patch. Please stop adding new patches. Use your original patch and updated per the documentation I provided.","accounts_in_message":[],"_revision_number":7},{"id":"0d94605a1af40aea007704bc06c085f33aad90e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 22:13:01.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/961c5dfe532145089336a1c041f11afd : FAILURE in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58beb70afaf1481dbdfccd320aeb28d4 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":7},{"id":"217d5c5faee65d4d2a203aadaebfb14e827c840c","author":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"date":"2020-10-05 05:16:29.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"eb33603acd589b7f97a4a817e8ab4d1ecf54f734","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 05:27:16.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9c6d0ad7d404440ad022809f7000b47 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/847e9e10d2ed4c6d95272eaa81e9388f : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":8},{"id":"f435e467db4714a6a6e150be84d39667524fa514","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-05 05:32:27.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/38/755338/8/check-not-vote/ext-glance-zuul/59214e9 : FAILURE in 15m 33s","accounts_in_message":[],"_revision_number":8},{"id":"cbc364ae446ab200461ee507c503daccf1513775","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-05 16:19:59.000000000","message":"Patch Set 8: Code-Review-1\n\n(10 comments)\n\nThis is still based on top of an abandoned patch. You will need to rebase this so it is directly on master, and include the changes you had tried to make in the abandoned patch.\n\nEasiest at this point is probably from the command line. Do something like the following:\n\ngit rebase -i HEAD~2\n[squash the two commits into one]\n[address comments to fix issues]\ngit add .\ngit commit --amend\n[use this patch\u0027s commit message, including Change-Id line]\ngit review\n\nThat should combine the two commits into one and have it based on master.","accounts_in_message":[],"_revision_number":8},{"id":"5eddaf11cff84b8f05a347ee57094ce7d372e12b","author":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"date":"2020-10-06 11:30:52.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"eb93023975dff6cdab11b153635d8b36732e0f5e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-06 11:33:40.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)\n\nStill based on top of an abandoned patch.","accounts_in_message":[],"_revision_number":9},{"id":"64f8e701dd9eac2d8846a1d8e41d2021508832a4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-06 11:34:18.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"76f9081f64a459e993649048d74ef503c000d592","author":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"date":"2020-10-06 13:29:24.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e222f53941bf2a1d46d60bb72d98040a11171839","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-06 13:37:03.000000000","message":"Patch Set 10:\n\nRebase this.","accounts_in_message":[],"_revision_number":10},{"id":"25fabec7126a7fbcd085d90204049e7c220d6a45","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-10-06 13:42:36.000000000","message":"Patch Set 10: Code-Review-1\n\nStill based on an abandoned patch, still looks to be half in spec form, still needs to be squashed into the original:\n\nhttps://review.opendev.org/#/c/754969/\n\nMaybe if you could join us on IRC (#openstack-glance channel) we could help you out with the procedures needed to squash all of these, as well as get the formatting fixed?\n\nhttps://wiki.openstack.org/wiki/IRC","accounts_in_message":[],"_revision_number":10},{"id":"c268884d712a4b907d70d2c465af2e361865a490","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-06 14:51:52.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/38/755338/10/check-not-vote/ext-glance-zuul/e25ba4f : FAILURE in 15m 03s","accounts_in_message":[],"_revision_number":10},{"id":"737df4407a64157136fa98357a77e111fb2a4d15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 20:07:24.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/055bf54d678f41239db4a864a60d25e9 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/712df6b8729847cfa284f5366f79ed7e : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":10},{"id":"ac0048896c4f190c446d5323c521b7e4a82205c4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-08 14:13:03.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"9bdb9da38086cf07aa343c2103db72b493f364cd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-08 14:29:07.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/38/755338/11/check-not-vote/ext-glance-zuul/bd83f50 : FAILURE in 15m 39s","accounts_in_message":[],"_revision_number":11},{"id":"2abfedad343736a5bf233c4b72ad1ef4758ceb8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 00:32:31.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fd92a5b7eef4f1cb9455ba76233ee42 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf9e81f08d4443ba94305b4f3ec281fc : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":11},{"id":"078d70367e6271c76a6bee4c8160173f49550270","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-29 14:31:30.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"00b989c9061883b1bc7bb28cca9cc63427e4b38c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-08 20:31:30.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"7d43e53334567f82508d8e46e8d14f0d7c69d7cd","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-08 20:32:37.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"f1637520ef3294ba3923768fdab44e1b5711e50d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 20:41:03.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a50691994c314278b9fd62b0b11dec20\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ced9ccdbb9b54daaa369aa3555af4d3b : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0b99f53fec4443286c48d619273231a : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":12},{"id":"43ecac7a88835c57eb53b4434f5ff6c866a62ba4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-02-09 14:35:21.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"09ac7761ad83070b3f98392216927326b505e80f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 14:36:00.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"83e10ec45783fc856f711f2007216cda606e23fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 15:50:47.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7204c172dc14496a2d6df4b7f399361\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb631f55d3ba4f0dbc09dbad03fba571 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4f9e9b294194634811ba04e4137b020 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":12},{"id":"bf890409d2e6a16943b5518c858e976b3a64a28d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 15:50:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"4e7a7ef739f9516d8a315a92671775a03316db97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 15:52:58.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65414063983b4a6f875a6f1de28e40d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3adf7d00ff8244ec9d79a5fe88f1d1e0 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"438758db2feeb59d75d6432b84e449df52ba84d6","revisions":{"a07f77d8eba8407de6b05d4127e846af674550d9":{"kind":"REWORK","_number":1,"created":"2020-09-30 16:12:20.000000000","uploader":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"ref":"refs/changes/38/755338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/755338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/755338/1"}}},"commit":{"parents":[{"commit":"2a82b03f5085989f4dfa2c231b65518ef0a43e2f","subject":"new file added","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a82b03f5085989f4dfa2c231b65518ef0a43e2f"}]}],"author":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-09-30 16:09:17.000000000","tz":330},"committer":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-09-30 16:10:58.000000000","tz":330},"subject":"Multihash Info is added","message":"Multihash Info is added\n\nos_hash_algo file is added in glance user and admin guides.\n\nChange-Id: Id78be3935998b9c5acdd0706393117e892e5ab59\nCloses-bug:#1871419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a07f77d8eba8407de6b05d4127e846af674550d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a07f77d8eba8407de6b05d4127e846af674550d9"}]},"branch":"refs/heads/master"},"cd87a96fd0b715dd49ddfc0acaf301aedaa6bdd0":{"kind":"REWORK","_number":2,"created":"2020-10-01 07:48:04.000000000","uploader":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"ref":"refs/changes/38/755338/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/755338/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/755338/2"}}},"commit":{"parents":[{"commit":"2a82b03f5085989f4dfa2c231b65518ef0a43e2f","subject":"new file added","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a82b03f5085989f4dfa2c231b65518ef0a43e2f"}]}],"author":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-09-30 16:09:17.000000000","tz":330},"committer":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-10-01 07:44:35.000000000","tz":330},"subject":"Multihash Info is added","message":"Multihash Info is added\n\nos_hash_algo file is added in glance user and admin guides.\n\nChange-Id: Id78be3935998b9c5acdd0706393117e892e5ab59\nCloses-bug:#1871419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cd87a96fd0b715dd49ddfc0acaf301aedaa6bdd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cd87a96fd0b715dd49ddfc0acaf301aedaa6bdd0"}]},"branch":"refs/heads/master"},"fc991e95cf70796db127cfe34dd59f4eaaf4e9db":{"kind":"REWORK","_number":3,"created":"2020-10-01 09:30:52.000000000","uploader":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"ref":"refs/changes/38/755338/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/755338/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/755338/3"}}},"commit":{"parents":[{"commit":"2a82b03f5085989f4dfa2c231b65518ef0a43e2f","subject":"new file added","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a82b03f5085989f4dfa2c231b65518ef0a43e2f"}]}],"author":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-09-30 16:09:17.000000000","tz":330},"committer":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-10-01 09:29:17.000000000","tz":330},"subject":"Multihash Info is added","message":"Multihash Info is added\n\nos_hash_algo file is added in glance user and admin guides.\n\nChange-Id: Id78be3935998b9c5acdd0706393117e892e5ab59\nCloses-bug:#1871419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc991e95cf70796db127cfe34dd59f4eaaf4e9db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc991e95cf70796db127cfe34dd59f4eaaf4e9db"}]},"branch":"refs/heads/master"},"882632791b23fe73e259dd7589c1a44e33df7dad":{"kind":"REWORK","_number":4,"created":"2020-10-01 09:37:55.000000000","uploader":{"_account_id":30905,"name":"Harshavardhan Metla","email":"harshavardhan.metla@tcs.com","username":"harsha24"},"ref":"refs/changes/38/755338/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/755338/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/755338/4"}}},"commit":{"parents":[{"commit":"2a82b03f5085989f4dfa2c231b65518ef0a43e2f","subject":"new file added","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a82b03f5085989f4dfa2c231b65518ef0a43e2f"}]}],"author":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-09-30 16:09:17.000000000","tz":330},"committer":{"name":"Harshavardhan Metla","email":"harshavardhan.metla@tcs.com","date":"2020-10-01 09:37:39.000000000","tz":0},"subject":"Multihash Info is added","message":"Multihash Info is added\n\nos_hash_algo file is added in glance user and admin guides.\n\nChange-Id: Id78be3935998b9c5acdd0706393117e892e5ab59\nCloses-bug:#1871419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/882632791b23fe73e259dd7589c1a44e33df7dad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/882632791b23fe73e259dd7589c1a44e33df7dad"}]},"branch":"refs/heads/master"},"29eef83a05dbb3d5e54b1ef047c465682b6956d8":{"kind":"REWORK","_number":5,"created":"2020-10-01 09:38:37.000000000","uploader":{"_account_id":30905,"name":"Harshavardhan Metla","email":"harshavardhan.metla@tcs.com","username":"harsha24"},"ref":"refs/changes/38/755338/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/755338/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/755338/5"}}},"commit":{"parents":[{"commit":"2a82b03f5085989f4dfa2c231b65518ef0a43e2f","subject":"new file added","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a82b03f5085989f4dfa2c231b65518ef0a43e2f"}]}],"author":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-09-30 16:09:17.000000000","tz":330},"committer":{"name":"Harshavardhan Metla","email":"harshavardhan.metla@tcs.com","date":"2020-10-01 09:38:31.000000000","tz":0},"subject":"Multihash Info is added","message":"Multihash Info is added\n\nos_hash_algo file is added in glance user and admin guides.\n\nChange-Id: Id78be3935998b9c5acdd0706393117e892e5ab59\nCloses-bug:#1871419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/29eef83a05dbb3d5e54b1ef047c465682b6956d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/29eef83a05dbb3d5e54b1ef047c465682b6956d8"}]},"branch":"refs/heads/master"},"a6a0a3cb2b2b03e27c5c2e1082704aed7171846b":{"kind":"REWORK","_number":6,"created":"2020-10-01 12:31:23.000000000","uploader":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"ref":"refs/changes/38/755338/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/755338/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/755338/6"}}},"commit":{"parents":[{"commit":"2a82b03f5085989f4dfa2c231b65518ef0a43e2f","subject":"new file added","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a82b03f5085989f4dfa2c231b65518ef0a43e2f"}]}],"author":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-09-30 16:09:17.000000000","tz":330},"committer":{"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","date":"2020-10-01 12:29:23.000000000","tz":0},"subject":"Multihash Info is added","message":"Multihash Info is added\n\nos_hash_algo file is added in glance user and admin guides.\n\nChange-Id: Id78be3935998b9c5acdd0706393117e892e5ab59\nCloses-bug:#1871419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6a0a3cb2b2b03e27c5c2e1082704aed7171846b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a6a0a3cb2b2b03e27c5c2e1082704aed7171846b"}]},"branch":"refs/heads/master"},"12423aecf20220101aa6ba8a0415d3a6a08fe059":{"kind":"REWORK","_number":7,"created":"2020-10-01 15:17:12.000000000","uploader":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"ref":"refs/changes/38/755338/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/755338/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/755338/7"}}},"commit":{"parents":[{"commit":"2a82b03f5085989f4dfa2c231b65518ef0a43e2f","subject":"new file added","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a82b03f5085989f4dfa2c231b65518ef0a43e2f"}]}],"author":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-09-30 16:09:17.000000000","tz":330},"committer":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-10-01 15:14:05.000000000","tz":330},"subject":"Multihash Info is added","message":"Multihash Info is added\n\nos_hash_algo file is added in glance user and admin guides.\n\nChange-Id: Id78be3935998b9c5acdd0706393117e892e5ab59\nCloses-bug:#1871419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/12423aecf20220101aa6ba8a0415d3a6a08fe059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/12423aecf20220101aa6ba8a0415d3a6a08fe059"}]},"branch":"refs/heads/master"},"04c0a8c528a0c029049276d170ca52829a74b059":{"kind":"REWORK","_number":8,"created":"2020-10-05 05:16:29.000000000","uploader":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"ref":"refs/changes/38/755338/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/755338/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/755338/8"}}},"commit":{"parents":[{"commit":"2a82b03f5085989f4dfa2c231b65518ef0a43e2f","subject":"new file added","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a82b03f5085989f4dfa2c231b65518ef0a43e2f"}]}],"author":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-09-30 16:09:17.000000000","tz":330},"committer":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-10-05 05:13:39.000000000","tz":330},"subject":"Multihash Info is added","message":"Multihash Info is added\n\nos_hash_algo file is added in glance user and admin guides.\n\nChange-Id: Id78be3935998b9c5acdd0706393117e892e5ab59\nCloses-bug:#1871419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/04c0a8c528a0c029049276d170ca52829a74b059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/04c0a8c528a0c029049276d170ca52829a74b059"}]},"branch":"refs/heads/master"},"af3d094b0cfde753eafd0a0b0411c453f3e79a7d":{"kind":"REWORK","_number":9,"created":"2020-10-06 11:30:52.000000000","uploader":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"ref":"refs/changes/38/755338/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/755338/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/755338/9"}}},"commit":{"parents":[{"commit":"2a82b03f5085989f4dfa2c231b65518ef0a43e2f","subject":"new file added","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a82b03f5085989f4dfa2c231b65518ef0a43e2f"}]}],"author":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-09-30 16:09:17.000000000","tz":330},"committer":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-10-06 11:28:34.000000000","tz":330},"subject":"Multihash Info is added","message":"Multihash Info is added\n\nos_hash_algo file is added in glance user and admin guides.\n\nChange-Id: Id78be3935998b9c5acdd0706393117e892e5ab59\nCloses-bug:#1871419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/af3d094b0cfde753eafd0a0b0411c453f3e79a7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/af3d094b0cfde753eafd0a0b0411c453f3e79a7d"}]},"branch":"refs/heads/master"},"26b8949618ff18d9f5feb036314ac91529bc6cf7":{"kind":"REWORK","_number":10,"created":"2020-10-06 13:29:24.000000000","uploader":{"_account_id":32410,"name":"Pragathi Marampudi","email":"pragathirj.marampudi@gmail.com","username":"pragathi93"},"ref":"refs/changes/38/755338/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/755338/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/755338/10"}}},"commit":{"parents":[{"commit":"2a82b03f5085989f4dfa2c231b65518ef0a43e2f","subject":"new file added","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a82b03f5085989f4dfa2c231b65518ef0a43e2f"}]}],"author":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-09-30 16:09:17.000000000","tz":330},"committer":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-10-06 13:28:20.000000000","tz":330},"subject":"Multihash Info is added","message":"Multihash Info is added\n\nos_hash_algo file is added in glance user and admin guides.\n\nChange-Id: Id78be3935998b9c5acdd0706393117e892e5ab59\nCloses-bug:#1871419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26b8949618ff18d9f5feb036314ac91529bc6cf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26b8949618ff18d9f5feb036314ac91529bc6cf7"}]},"branch":"refs/heads/master"},"19166b92474473c231abba75f34b52e4266ac66f":{"kind":"REWORK","_number":11,"created":"2020-10-08 14:13:03.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/38/755338/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/755338/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/755338/11"}}},"commit":{"parents":[{"commit":"b49c0424672ac8191adda5018957a7317fae9de1","subject":"Adjust jobs for devstack WSGI mode default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b49c0424672ac8191adda5018957a7317fae9de1"}]}],"author":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-09-30 07:51:03.000000000","tz":330},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-10-08 14:11:20.000000000","tz":-300},"subject":"Add multihash info in glance documentation","message":"Add multihash info in glance documentation\n\nNew docs added for os_hash_algo in user guide and admin guide.\n\nChange-Id: Id78be3935998b9c5acdd0706393117e892e5ab59\nCloses-bug:#1871419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/19166b92474473c231abba75f34b52e4266ac66f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/19166b92474473c231abba75f34b52e4266ac66f"}]},"branch":"refs/heads/master"},"438758db2feeb59d75d6432b84e449df52ba84d6":{"kind":"REWORK","_number":12,"created":"2023-02-08 20:31:30.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/38/755338/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/755338/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/755338/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/755338/12"}}},"commit":{"parents":[{"commit":"0f8e5ef01748fe75d4752d406ff2a05c419f7fb2","subject":"Merge \"Update migration constant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f8e5ef01748fe75d4752d406ff2a05c419f7fb2"}]}],"author":{"name":"pragathi93","email":"pragathirj.marampudi@gmail.com","date":"2020-09-30 07:51:03.000000000","tz":330},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-02-08 20:30:44.000000000","tz":60},"subject":"Add multihash info in glance documentation","message":"Add multihash info in glance documentation\n\nNew docs added for os_hash_algo in user guide and admin guide.\n\nChange-Id: Id78be3935998b9c5acdd0706393117e892e5ab59\nCloses-bug:#1871419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/438758db2feeb59d75d6432b84e449df52ba84d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/438758db2feeb59d75d6432b84e449df52ba84d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
