)]}'
{"id":"openstack%2Fglance~757674","triplet_id":"openstack%2Fglance~master~Ieb3131baab3d2382055388abc5b669304fadd478","project":"openstack/glance","branch":"master","topic":"six","hashtags":[],"change_id":"Ieb3131baab3d2382055388abc5b669304fadd478","subject":"[Part1] Remove six","status":"ABANDONED","created":"2020-10-13 05:46:57.000000000","updated":"2022-02-16 21:24:27.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ed8ca7fbccc8466d01951064dd8086f65867d785","_number":757674,"virtual_id_number":757674,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-14 16:07:09.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-13 11:14:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-13 11:14:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 16:07:09.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a59c8c4d62503f0bbf092e27bab97b454d30698","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-13 05:46:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"803e6c150852bbb9e3c437b46015283fd8992c43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 06:49:43.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/91a8ff6d7374418a8f71a9b44765fc37 : FAILURE in 1h 01m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7bde4b07f36a42e1951fb4c525907958 : FAILURE in 32m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e4ab98a12ae42c7ab4781a3936880f3 : FAILURE in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea4ceba19fad4fe786ef919541ba6928 : FAILURE in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e521d5ba98449a4b02dd1ba992f75d6 : FAILURE in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72329a24acf64b6689c105de78750c0e : FAILURE in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11dd3bbb51504fd19e6b83782fa116ad : FAILURE in 4m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3733020dd58844a4a32469a1138f3dcc : FAILURE in 10m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dcc0e200514b44dd9265b1d1b33ce4dd : FAILURE in 11m 16s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ebf5db1ae7534b13af8f0f86908107ea : SUCCESS in 3m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fd1c5af6badb4c3495fd3c0580e148ac : FAILURE in 31m 21s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ef58dc31e4594562be8b04facc94dedc : POST_FAILURE in 30m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cc946255532540829314a6a491e33bb4 : POST_FAILURE in 34m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/20081b0e4aa9436b9c802f9d2c0c7d76 : POST_FAILURE in 30m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd732c747da74ddb9a6a59654e7351c8 : FAILURE in 30m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/670227b507c246979c02593c4a59ae60 : FAILURE in 31m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b3e930d54f73058be31d213f89318e05db78a356","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-13 06:59:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ba45e912f6c7c45793deda5f26b2dae50d91fdab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 11:14:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b0ef507b88e949f894e7531acdfe3864 : SUCCESS in 1h 02m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/15fb8ffb57e14de198f9fb0b31003bdf : SUCCESS in 1h 09m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f85b4561d5d54c338299a439f677901c : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28ce1f75c89a4d1a8856383706f5c9f2 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53f28df43af446c6a7b8bafd9971434c : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ad00bd1b7bc44c8b98e5a73f141a603 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62d8cdea728b4b5fb22753565865ddc8 : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/567ca7c0430246d49af00790f438f9e9 : SUCCESS in 7m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dc0843d11b554b77a35ede342ea727df : SUCCESS in 8m 56s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/30088bcd48074cb8b4b5738634492824 : SUCCESS in 6m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3e05cd7410e8493eb9d5cc8365d1a62c : SUCCESS in 1h 06m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/822071cadeb94bdeaa7f68ffae2731c9 : SUCCESS in 1h 18m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5d6a17002d2a4e8c9ade8b3b07945dab : FAILURE in 1h 13m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5f93f660c1b84cc2a8c49bb21e32c155 : SUCCESS in 1h 15m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7d1891c15c847f2ab972b6a4d9331fd : SUCCESS in 1h 02m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7d43204d103844dd8175c995f0d44e46 : SUCCESS in 1h 09m 50s","accounts_in_message":[],"_revision_number":2},{"id":"06a189fb65c3dd94ef18b0af1e901445e1aaa67f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-14 16:07:09.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/74/757674/2/check-not-vote/ext-glance-zuul/b9af92e : FAILURE in 15m 48s","accounts_in_message":[],"_revision_number":2},{"id":"4b1defca70cd5659d22462e818e2fcf6ed5a7c34","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-05-04 21:20:01.000000000","message":"Topic set to six","accounts_in_message":[],"_revision_number":2},{"id":"ed8ca7fbccc8466d01951064dd8086f65867d785","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-16 21:24:27.000000000","message":"Abandoned\n\nWe removed six in another series of commits by Stephen Finucane.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"909774003c3b39d53d8958f23c88d6d2093533da","revisions":{"8922f19383902f399d198ce0f9c9fc2b71ade6b2":{"kind":"REWORK","_number":1,"created":"2020-10-13 05:46:57.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/74/757674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/757674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/757674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/757674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/757674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/757674/1"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 05:40:56.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 05:45:24.000000000","tz":480},"subject":"[Part1] Remove six We don\u0027t need this in a Python 3-only world.","message":"[Part1] Remove six\nWe don\u0027t need this in a Python 3-only world.\n\nRemove six in follows:\n1. glance/api/middleware/cache.py\n2. glance/api/v2/image_actions.py\n3. glance/api/v2/image_data.py\n4. glance/api/v2/image_members.py\n\nChange-Id: Ieb3131baab3d2382055388abc5b669304fadd478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8922f19383902f399d198ce0f9c9fc2b71ade6b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8922f19383902f399d198ce0f9c9fc2b71ade6b2"}]},"branch":"refs/heads/master"},"909774003c3b39d53d8958f23c88d6d2093533da":{"kind":"REWORK","_number":2,"created":"2020-10-13 06:59:02.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/74/757674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/74/757674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/757674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/757674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/74/757674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/74/757674/2"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 05:40:56.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 06:57:36.000000000","tz":480},"subject":"[Part1] Remove six","message":"[Part1] Remove six\n\nWe don\u0027t need this in a Python 3-only world.\nRemove six in follows:\n1. glance/api/middleware/cache.py\n2. glance/api/v2/image_actions.py\n3. glance/api/v2/image_data.py\n4. glance/api/v2/image_members.py\n\nChange-Id: Ieb3131baab3d2382055388abc5b669304fadd478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/909774003c3b39d53d8958f23c88d6d2093533da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/909774003c3b39d53d8958f23c88d6d2093533da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
