)]}'
{"id":"openstack%2Fglance~757678","triplet_id":"openstack%2Fglance~master~I5fc3065295201720b7cfbc71077d499434ab5cf0","project":"openstack/glance","branch":"master","topic":"six","hashtags":[],"change_id":"I5fc3065295201720b7cfbc71077d499434ab5cf0","subject":"[Part4] Remove six","status":"ABANDONED","created":"2020-10-13 05:59:16.000000000","updated":"2022-02-16 21:23:11.000000000","total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cd92a9a7fd70e4a24cf866c19803689f424aeb1f","_number":757678,"virtual_id_number":757678,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-14 23:41:08.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-14 04:39:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 04:39:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 23:41:08.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"996bdc42b8a4bbaa9ee5fe0daceba62c69e58952","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-13 05:59:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b9ca7a1ed1e62eec769fc7def8de3bb365da32c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 07:06:33.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/239ce637b72641fe8e76edd58d7aea86 : FAILURE in 1h 04m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cdae49fb09094e70bf222823b3bd553e : FAILURE in 28m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50a1bf6b0ffa42588944c4d649f86baa : FAILURE in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50f17a0bfa7b44abb8497f1047d7bed2 : FAILURE in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/031f14d6a55641e59771db41c2e36d9b : FAILURE in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5786b04c967410094fdb67c605896f2 : FAILURE in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d159a7daac4448291f5c05d38b0a109 : FAILURE in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8888a35975af4b01bf5c33b268489b4c : FAILURE in 4m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5d5b213daa8344c7be8daf29f93063f2 : FAILURE in 5m 11s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e479e8a1a89443118194b181422ed013 : SUCCESS in 4m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3d4b6db34ed54dda910bfad857c46039 : FAILURE in 31m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f4991c70b7aa42d283f9f1d35299e27c : POST_FAILURE in 37m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1d09d65e92ab4ba9bdb15d896ad3c5be : POST_FAILURE in 32m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/aa69b778a3d24c2e93c7e1d5f4273c37 : POST_FAILURE in 33m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d909c4b82c54fb9acfe493e5f750a49 : FAILURE in 37m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0d024b63da1244588e9df6820042ae78 : FAILURE in 31m 52s","accounts_in_message":[],"_revision_number":1},{"id":"36db7e3fdd9cbbba250c4ffad59e6164d3fa7a97","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-13 07:34:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"014b568d7b0d5b1dc1c74236e48ddb65c58ec8b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 12:20:53.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/1b9ffbeabe3f46ceb722e73efb3471ab : FAILURE in 59m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e0a013b031ed460fb940a7e869634289 : FAILURE in 29m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ac915291f9944cf80ac49a996622e7b : FAILURE in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0d32ccf90394ef5932a9ef8c9dde203 : FAILURE in 11m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d8f01ad311c498780cddf0b013204ef : FAILURE in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43ba9657045e45508025438f812fda7b : FAILURE in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5abb9727ee2f4a6683d47b08f94db7e3 : FAILURE in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d9a680aad1c448528cae664b8f9db65d : FAILURE in 4m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/47b3cb38747d4d949c73644d72bb9672 : FAILURE in 4m 51s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9f9f501e426e4a259d2c1004dc378d4f : SUCCESS in 6m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/76556128f6884703925b9b3d934ace21 : FAILURE in 27m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f7e68712ca724891b64dc3117dff9cd4 : POST_FAILURE in 33m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a9709b4ba04e4a199231465b43f28fcf : POST_FAILURE in 28m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6391abd600514e6b987ada787e5fd3d5 : POST_FAILURE in 32m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ed89f5710cd4429aa57b68355631f24 : FAILURE in 33m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e01538764cac446d8e010eaf0dd86767 : FAILURE in 37m 27s","accounts_in_message":[],"_revision_number":2},{"id":"2984297847017d93876d451083d9f7beb4ee9cf8","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-13 14:50:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9d15f0c9ce3769dce6a514af505b2d90e0fb989d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 20:02:18.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/1b160c676db840e395c101d6feb76642 : FAILURE in 56m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b892a2e61704e8ebd70c017d55baf1a : FAILURE in 30m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4f7176d03664ec5a8b3f5d03d7f9b15 : FAILURE in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df5d3680c93145159be2341fb609704e : FAILURE in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20f6a46a432a44cc8264e5faa877e8e2 : FAILURE in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/038ba52ffd3a490dab3cab0b660289c5 : FAILURE in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89a3e6cdcc64b7db2e1d33d969bfb59 : FAILURE in 5m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2c5e0a3b0e8a458081df511a9504cb17 : FAILURE in 4m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/469b02f355824970a338b770e138eefd : FAILURE in 5m 41s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/40cf739c17234a47887f759935b4874a : SUCCESS in 4m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9324565326b5455ca0a78c6ca8c6ea5e : FAILURE in 25m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/010b213f740d4da7b436f9376f53fcce : POST_FAILURE in 34m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1531ebb76244483d8af95b0963f34d2d : POST_FAILURE in 31m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1cc069f09c0041f1a85f41759cfa4359 : POST_FAILURE in 35m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bd620ed04ce40d68a73346221ec3e44 : FAILURE in 33m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ac6bcd13aa60434f83eb430c48407fdf : FAILURE in 33m 17s","accounts_in_message":[],"_revision_number":3},{"id":"51971d30fc2058cba78d903dd35d0c3aa78f37aa","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-13 22:30:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e373fc4b4e1336148fc6600f175786b412a8d4a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 23:31:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/3a0f696789ae4d9c962fa22eac41eca7 : FAILURE in 55m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/04373d3fc2f14c1981b45faafb0dfab0 : FAILURE in 28m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25c85d60127f4d6fb54a25018320d5cb : FAILURE in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa0becece6f24a9c823705b99f64086e : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8df305aa6e6b4905974b135f77e7e613 : FAILURE in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a656801f2af34ee0a2c33afe35184f22 : FAILURE in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07bbdb7db7af4360b8a73466046e2ca3 : FAILURE in 4m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65b139ee47ae4af3b2b65269a8f7bfb8 : FAILURE in 4m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/28ab8b6499f14cf2ba506d511ad7468b : FAILURE in 4m 13s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/18c39c5209fb4b0388eeb569c86dae9a : SUCCESS in 6m 51s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/65ced3e1afb04c378a1464df2061c6d4 : FAILURE in 32m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/85ed3adb06dd461bb2f5d40ec71fb864 : POST_FAILURE in 24m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/299ab4ed4e024547932bdcfc0e09a1b5 : POST_FAILURE in 45m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ecc9f0a1029b4d91a15f9c9e5f4cb2c1 : POST_FAILURE in 36m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0755cc4ff2674787949870163e438d8d : FAILURE in 30m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/284b3b2894cf4c0d8757087cfb238fed : FAILURE in 32m 01s","accounts_in_message":[],"_revision_number":4},{"id":"ebba5188836e70e7ad787a5c084716574f3c6755","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-14 00:49:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b7fa35f240135044208c6cfcca7aa1228806bd1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 02:01:24.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/39b476c067b747c4a2433e35ebb01796 : FAILURE in 1h 11m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1c3d159b1d764b67adc83fb75316cabe : FAILURE in 46m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a856e01c9eef4ffe8655945e78db31a9 : FAILURE in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1603585c70f4a168612743fb2a838ea : FAILURE in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/caf79bbf1aa44da082f3a77a5e8810d5 : FAILURE in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68ad32151f4e46f0b2f64527526ddd5c : FAILURE in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05ea87bb63d2471cb30bdcb5d98e3fa8 : FAILURE in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5d463ea66d59456e88b43e81fdb71ccb : FAILURE in 6m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/35bbf429030f4f70a2b4d2285255dcd4 : FAILURE in 4m 59s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/52e68265112e444e96cab2fd5e9dd23a : SUCCESS in 4m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e09220f5eba949959ad37261ec3f8a63 : FAILURE in 31m 15s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ad5034d8b862404091176fa7504be53a : POST_FAILURE in 29m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/70e9429820794abcbbf2cfbe9b1fab0a : POST_FAILURE in 35m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5fdc06f014cc455a9b92a848bd7757c2 : POST_FAILURE in 34m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae252d4ee51049f79239a7773b72383b : FAILURE in 31m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aa1d2872cc67412d86e6ec7535d0a68e : FAILURE in 28m 41s","accounts_in_message":[],"_revision_number":5},{"id":"f9a846032b58fc45a7376077f408b8dfd5186be8","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-14 03:14:08.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3b66460928a8c51c2fcd5ee917a3036d716561ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 04:39:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/44fb726fe6c6419bb5d44512b95285fd : SUCCESS in 58m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dc5ede6304504f698af6cfa01c9b742a : SUCCESS in 1h 15m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9843fe9889b04029a1ba89380faa132f : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa1352e66d2941e4b202985964341cba : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9b72f7a836449dbbb3e8ba92e088288 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e65654ac2ee94b6bb14219775e0f6d27 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96df6b2da0694f19bb938c864e49139d : SUCCESS in 5m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/484fef0d1df94850833f51ee68cd8091 : SUCCESS in 7m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/67221fe9b4804fa0ad53f523a81ddb74 : SUCCESS in 10m 03s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/20350768174c4b8493a219b8a9ed3948 : SUCCESS in 4m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/deb497c0d71d49b18498423b603ab0c4 : SUCCESS in 1h 07m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fe3bd70b84a14fb79c337512641517e0 : SUCCESS in 1h 13m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7e5f4bbbcde245eab66084134b3490b4 : SUCCESS in 1h 12m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5002884281f248fba713f0d4ab4a1f5d : FAILURE in 1h 07m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b693101566d468e80db660537f6f6ab : SUCCESS in 1h 02m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eb1966632ae24e30b273426d1e3627af : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":6},{"id":"7e3f4b54ec168576148cdd1b6921210b559f080b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-14 23:41:08.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/78/757678/6/check-not-vote/ext-glance-zuul/71f9a80 : FAILURE in 15m 13s","accounts_in_message":[],"_revision_number":6},{"id":"dab699c36bc07fd29513860c7552d0fc1b6c56d1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-05-04 21:20:45.000000000","message":"Topic set to six","accounts_in_message":[],"_revision_number":6},{"id":"cd92a9a7fd70e4a24cf866c19803689f424aeb1f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-16 21:23:11.000000000","message":"Abandoned\n\nWe removed six in another series of commits by Stephen Finucane.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e056e2f9b078fea74639e99142ea4430c541a8e0","revisions":{"8b7a31b90d7d2f7429b1b3fb67285125f40779d0":{"kind":"REWORK","_number":1,"created":"2020-10-13 05:59:16.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/78/757678/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/757678/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/757678/1"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 05:53:39.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 05:58:53.000000000","tz":480},"subject":"[Part4] Remove six","message":"[Part4] Remove six\n\nWe don\u0027t need this in a Python 3-only world.\nRemove six in follows:\n\n1. glance/api/v2/tasks.py\n2. glance/api/versions.py\n3. glance/async_/flows/api_image_import.py\n\nChange-Id: I5fc3065295201720b7cfbc71077d499434ab5cf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8b7a31b90d7d2f7429b1b3fb67285125f40779d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8b7a31b90d7d2f7429b1b3fb67285125f40779d0"}]},"branch":"refs/heads/master"},"d2430cdcd95196253d25b792987a11c668a91cf0":{"kind":"REWORK","_number":2,"created":"2020-10-13 07:34:03.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/78/757678/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/757678/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/757678/2"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 05:53:39.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 07:32:57.000000000","tz":480},"subject":"[Part4] Remove six","message":"[Part4] Remove six\n\nWe don\u0027t need this in a Python 3-only world.\nRemove six in follows:\n\n1. glance/api/v2/tasks.py\n2. glance/api/versions.py\n3. glance/async_/flows/api_image_import.py\n\nChange-Id: I5fc3065295201720b7cfbc71077d499434ab5cf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2430cdcd95196253d25b792987a11c668a91cf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2430cdcd95196253d25b792987a11c668a91cf0"}]},"branch":"refs/heads/master"},"f454d6fe10505ef3f1594bdd9d3231aa76dc2ead":{"kind":"REWORK","_number":3,"created":"2020-10-13 14:50:30.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/78/757678/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/757678/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/757678/3"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 05:53:39.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 14:49:42.000000000","tz":480},"subject":"[Part4] Remove six","message":"[Part4] Remove six\n\nWe don\u0027t need this in a Python 3-only world.\nRemove six in follows:\n\n1. glance/api/v2/tasks.py\n2. glance/api/versions.py\n3. glance/async_/flows/api_image_import.py\n\nChange-Id: I5fc3065295201720b7cfbc71077d499434ab5cf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f454d6fe10505ef3f1594bdd9d3231aa76dc2ead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f454d6fe10505ef3f1594bdd9d3231aa76dc2ead"}]},"branch":"refs/heads/master"},"a3342a3b88243182ac5a8632f9228a1751e39750":{"kind":"REWORK","_number":4,"created":"2020-10-13 22:30:46.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/78/757678/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/757678/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/757678/4"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 05:53:39.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 22:30:18.000000000","tz":480},"subject":"[Part4] Remove six","message":"[Part4] Remove six\n\nWe don\u0027t need this in a Python 3-only world.\nRemove six in follows:\n\n1. glance/api/v2/tasks.py\n2. glance/api/versions.py\n3. glance/async_/flows/api_image_import.py\n\nChange-Id: I5fc3065295201720b7cfbc71077d499434ab5cf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a3342a3b88243182ac5a8632f9228a1751e39750"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a3342a3b88243182ac5a8632f9228a1751e39750"}]},"branch":"refs/heads/master"},"7884cc963314cc5ebf9fcb6e6a8241b0a118c7cf":{"kind":"REWORK","_number":5,"created":"2020-10-14 00:49:09.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/78/757678/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/757678/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/757678/5"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 05:53:39.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-14 00:48:51.000000000","tz":480},"subject":"[Part4] Remove six","message":"[Part4] Remove six\n\nWe don\u0027t need this in a Python 3-only world.\nRemove six in follows:\n\n1. glance/api/v2/tasks.py\n2. glance/api/versions.py\n3. glance/async_/flows/api_image_import.py\n\nChange-Id: I5fc3065295201720b7cfbc71077d499434ab5cf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7884cc963314cc5ebf9fcb6e6a8241b0a118c7cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7884cc963314cc5ebf9fcb6e6a8241b0a118c7cf"}]},"branch":"refs/heads/master"},"e056e2f9b078fea74639e99142ea4430c541a8e0":{"kind":"REWORK","_number":6,"created":"2020-10-14 03:14:08.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/78/757678/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/757678/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/757678/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/757678/6"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 05:53:39.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-14 03:13:09.000000000","tz":480},"subject":"[Part4] Remove six","message":"[Part4] Remove six\n\nWe don\u0027t need this in a Python 3-only world.\nRemove six in follows:\n\n1. glance/api/v2/tasks.py\n2. glance/api/versions.py\n3. glance/async_/flows/api_image_import.py\n\nChange-Id: I5fc3065295201720b7cfbc71077d499434ab5cf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e056e2f9b078fea74639e99142ea4430c541a8e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e056e2f9b078fea74639e99142ea4430c541a8e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
