)]}'
{"id":"openstack%2Fglance~757686","triplet_id":"openstack%2Fglance~master~Ic3db42643e2860d7443885c5ee46e5af284f8e2f","project":"openstack/glance","branch":"master","topic":"six","hashtags":[],"change_id":"Ic3db42643e2860d7443885c5ee46e5af284f8e2f","subject":"[Part9] Remove six","status":"ABANDONED","created":"2020-10-13 06:15:08.000000000","updated":"2022-02-16 21:24:32.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"515b11c217a78b14af360e369fc13789bc0ac988","_number":757686,"virtual_id_number":757686,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-14 17:43:19.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-13 13:30:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-13 13:30:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 17:43:19.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1016fa313e2b54be10a39f80dafc7c699100fed6","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-13 06:15:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ff939bc29d386c8518baccd5401a5d93c7b4b07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 08:52:28.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d9222b0bd5314289a679d4fce472740e : FAILURE in 42m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c2b8223ee9b341a7af2ddcb0721ac0ff : FAILURE in 28m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8360a8d31cbc4f949704af599a95bab1 : FAILURE in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b563ada2ae0b4578ac3a7c3c2ff411fb : FAILURE in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a44d14d06b6409293ec5a54fcbce113 : FAILURE in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b13871b611247c1a07a4b3ce63aec9c : FAILURE in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38c9a6e935d844acacbbe5f2e2b66061 : FAILURE in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b0ec01bf52e345f289bb626d2c4f38b8 : FAILURE in 5m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b164e27c92c048fa9ec5a8e515ce661e : FAILURE in 6m 19s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/032d41105d524ef2afadf596594f1f6a : SUCCESS in 6m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/210c3412678b4ff4bd73db465cd676f5 : FAILURE in 27m 41s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1fec344c7d444f009eab7849526fb7ff : POST_FAILURE in 23m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f1decd5b6827499da7f539f2f54fa57a : POST_FAILURE in 31m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9f9ad23506984d45898790adc03e2196 : POST_FAILURE in 24m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbb52abd423d4551a21d5b137ca8f407 : FAILURE in 23m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/37aa9c994f0c4a30817f257209e7cabc : FAILURE in 27m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5a708cc5b0a605147833b1fc37fcadd7e5085cce","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-13 09:02:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"58950eee98d1650aebdebb3e5d2fd2c5412712cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 13:30:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/165dc394a3234c6886f42a55cf496bdb : SUCCESS in 1h 04m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0fe352362c614c79915bea8141676195 : SUCCESS in 1h 23m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a32d98233554b59ba601b37b094919e : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84d3a3394b8e4b2a9ee29f1061157d22 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/733102cf38034ceeb3fdcfec1927c335 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37534de96df2471cac4a28740d6a5589 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e13d8c1d9ebe470e8e899f676d7e4f16 : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aaa240ed1ce845d280d6b4c429234641 : SUCCESS in 6m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/415aae31f3854d99a2f71ddf94e4d9d8 : SUCCESS in 6m 58s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0e41db9d04a048339763beb050f8985a : SUCCESS in 4m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e7a9f0ab50ab445b8c212f06afdd81a3 : SUCCESS in 1h 09m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/28f0cf0269a443d196ef7dea5e1407a3 : SUCCESS in 1h 09m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/aa2117a2b00646b386324cc19f8a14e4 : SUCCESS in 1h 15m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/46131d9d5ef74624b2606dbe3ce6f453 : FAILURE in 1h 15m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f0710f30d9f42e3be75d9688b3960e6 : SUCCESS in 1h 03m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/be90502d68e3489891f6486d29096d24 : SUCCESS in 1h 06m 46s","accounts_in_message":[],"_revision_number":2},{"id":"d49a3d0092e3406f3b50b9aea99bd360b9dc42a3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-14 17:43:19.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/86/757686/2/check-not-vote/ext-glance-zuul/dea175f : FAILURE in 7m 36s","accounts_in_message":[],"_revision_number":2},{"id":"11a71cf65122d58ee2178a9c1469ccfccf8bacc3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-05-04 21:19:51.000000000","message":"Topic set to six","accounts_in_message":[],"_revision_number":2},{"id":"515b11c217a78b14af360e369fc13789bc0ac988","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-16 21:24:32.000000000","message":"Abandoned\n\nWe removed six in another series of commits by Stephen Finucane.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f2936d7eaf56a4a1c0f061c0123ce519f14879fc","revisions":{"87fe9605999eaf4bb2e5dbae21839d324e6a85eb":{"kind":"REWORK","_number":1,"created":"2020-10-13 06:15:08.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/86/757686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/86/757686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/757686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/757686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/757686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/86/757686/1"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 06:12:33.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 06:14:45.000000000","tz":480},"subject":"[Part9] Remove six","message":"[Part9] Remove six\n\nWe don\u0027t need this in a Python 3-only world.\n\nRemove six in follows:\n1.  glance/common/wsgi.py\n2.  glance/db/simple/api.py\n3.  glance/db/sqlalchemy/api.py\n\nChange-Id: Ic3db42643e2860d7443885c5ee46e5af284f8e2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/87fe9605999eaf4bb2e5dbae21839d324e6a85eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/87fe9605999eaf4bb2e5dbae21839d324e6a85eb"}]},"branch":"refs/heads/master"},"f2936d7eaf56a4a1c0f061c0123ce519f14879fc":{"kind":"REWORK","_number":2,"created":"2020-10-13 09:02:26.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/86/757686/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/86/757686/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/757686/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/757686/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/757686/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/86/757686/2"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 06:12:33.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 09:01:25.000000000","tz":480},"subject":"[Part9] Remove six","message":"[Part9] Remove six\n\nWe don\u0027t need this in a Python 3-only world.\n\nRemove six in follows:\n1.  glance/common/wsgi.py\n2.  glance/db/simple/api.py\n3.  glance/db/sqlalchemy/api.py\n\nChange-Id: Ic3db42643e2860d7443885c5ee46e5af284f8e2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2936d7eaf56a4a1c0f061c0123ce519f14879fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2936d7eaf56a4a1c0f061c0123ce519f14879fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
