)]}'
{"id":"openstack%2Fglance~757691","triplet_id":"openstack%2Fglance~master~I93b06b535314398fbea34c3899df70a81baff589","project":"openstack/glance","branch":"master","topic":"six","hashtags":[],"change_id":"I93b06b535314398fbea34c3899df70a81baff589","subject":"[Part13] Remove six","status":"ABANDONED","created":"2020-10-13 06:23:14.000000000","updated":"2022-02-16 21:23:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bbcedb2125fdc4c9260c6777de13364ac97bd290","_number":757691,"virtual_id_number":757691,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-14 14:39:35.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-13 10:11:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-13 10:11:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 14:39:35.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b6d431314f2bee94463b7174cbbce96d36349b54","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-13 06:23:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43d25154a85e3bd500f0aecdc22e302d630b3e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 10:11:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bfa136aacf14138b91f8c5a723863a4 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e386d946a66d4015a82aeed1645fdfc3 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffe33948f9544c69aefc364e756c46f2 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60ec08276e824405b3ecdc86f38182ea : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48062d8e0a2646c6814e23d115453882 : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/68b059038d8b4f87a24b473f8ed10ab4 : SUCCESS in 9m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b0d17c1fb1f84db797618beff6250c5d : SUCCESS in 8m 19s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/343ba228dc1c4c1c96aa15f17ede4eea : FAILURE in 11m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a1629ecb4dc441198ddbb496ba21a11d : SUCCESS in 1h 08m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7e51f9501afe0607d8b4bc2899879e74d1515577","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-14 14:39:35.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/91/757691/1/check-not-vote/ext-glance-zuul/f422bea : FAILURE in 16m 20s","accounts_in_message":[],"_revision_number":1},{"id":"3ad8884c3da93eef99b489f63e11b9db1868aca0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-05-04 21:20:11.000000000","message":"Topic set to six","accounts_in_message":[],"_revision_number":1},{"id":"bbcedb2125fdc4c9260c6777de13364ac97bd290","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-16 21:23:26.000000000","message":"Abandoned\n\nWe removed six in another series of commits by Stephen Finucane.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4cf49ad256eb750e6f6cd97baf2e6e3ddfa9bc77","revisions":{"4cf49ad256eb750e6f6cd97baf2e6e3ddfa9bc77":{"kind":"REWORK","_number":1,"created":"2020-10-13 06:23:14.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/91/757691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/757691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/757691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/757691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/757691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/757691/1"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 06:20:14.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 06:22:52.000000000","tz":480},"subject":"[Part13] Remove six","message":"[Part13] Remove six\n\nWe don\u0027t need this in a Python 3-only world.\nRemove six in:\nglance/tests/functional/v2/test_images.py\n\nChange-Id: I93b06b535314398fbea34c3899df70a81baff589\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4cf49ad256eb750e6f6cd97baf2e6e3ddfa9bc77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4cf49ad256eb750e6f6cd97baf2e6e3ddfa9bc77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
