)]}'
{"id":"openstack%2Fglance~757695","triplet_id":"openstack%2Fglance~master~I059b848f3f9a929541060a399c8a7a3c377374d8","project":"openstack/glance","branch":"master","topic":"six","hashtags":[],"change_id":"I059b848f3f9a929541060a399c8a7a3c377374d8","subject":"[Part14] Remove six","status":"ABANDONED","created":"2020-10-13 06:27:56.000000000","updated":"2022-02-16 21:23:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c9038e162f2a117ffdc24b262e83ad0473c8f449","_number":757695,"virtual_id_number":757695,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-14 14:53:14.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-13 10:19:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-13 10:19:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 14:53:14.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1989e99d1ab720ec4654ad0827822b63926801d5","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-13 06:27:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"555fa6ed6568bc6bd1123fdad2a41f0dfb93b4e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 10:19:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/676797cfa93d4617a518a22afa6b14e5 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c3ea75aec414da9992a8bf5c187de57 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49f82396cca148bd9deb4b4fffdecad1 : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/636f9eb2bcd04029a1b59eed0399558f : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b88d17446bfa42b5b13fc24d24574d12 : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3630afc587604c228a4d1a35ecd16c72 : SUCCESS in 8m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bd8eaba866e941db9c9625ac8499c53b : SUCCESS in 9m 41s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d59d7e4c0f824eecb1e630b4bb5d2452 : SUCCESS in 4m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d8cff2cf5c024503b6148c6a0f9d6550 : SUCCESS in 1h 14m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f8950cc2da6dc41d420e3dc8be123cdf4267b925","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-14 14:53:14.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/95/757695/1/check-not-vote/ext-glance-zuul/5ce1e52 : FAILURE in 17m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fe408ef6e7f3507ccdb91287c3161fb2672dac64","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-05-04 21:20:06.000000000","message":"Topic set to six","accounts_in_message":[],"_revision_number":1},{"id":"c9038e162f2a117ffdc24b262e83ad0473c8f449","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-16 21:23:32.000000000","message":"Abandoned\n\nWe removed six in another series of commits by Stephen Finucane.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c472da5cc45f24fb070234513656a2c899e7a28d","revisions":{"c472da5cc45f24fb070234513656a2c899e7a28d":{"kind":"REWORK","_number":1,"created":"2020-10-13 06:27:56.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/95/757695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/757695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/757695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/757695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/757695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/757695/1"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 06:25:32.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-13 06:27:33.000000000","tz":480},"subject":"[Part14] Remove six","message":"[Part14] Remove six\n\nWe don\u0027t need this in a Python 3-only world.\nRemove six in follow:\nglance/tests/functional/v2/test_images.py\n\nChange-Id: I059b848f3f9a929541060a399c8a7a3c377374d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c472da5cc45f24fb070234513656a2c899e7a28d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c472da5cc45f24fb070234513656a2c899e7a28d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
