)]}'
{"id":"openstack%2Fglance~757911","triplet_id":"openstack%2Fglance~master~Iacaddd9ae6a9af3d34b5e58d604badb01360fae7","project":"openstack/glance","branch":"master","topic":"hacking","hashtags":[],"change_id":"Iacaddd9ae6a9af3d34b5e58d604badb01360fae7","subject":"Address new hacking enforcement issues","status":"ABANDONED","created":"2020-10-13 16:44:36.000000000","updated":"2021-07-26 13:39:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6c4a3bdb1b79392883c2460172b43729bc1838ee","_number":757911,"virtual_id_number":757911,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-12-05 19:21:56.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"date":"2020-12-04 19:26:47.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-13 22:16:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"disliked":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":-1,"date":"2020-12-07 14:52:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-10-13 16:51:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-12-07 14:52:55.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-13 16:51:31.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-10-13 22:16:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-21 13:04:14.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-07 14:52:55.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"146bcde78c7e174274d065eba82332df79309917","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-13 16:44:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a738807e0362f9a04c8bc0fb8e36c15cbe6965b8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-10-13 16:51:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6ac55d39bc4f289ce608b97bfc1576e2fafe5867","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 22:16:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9ee8e0f085724daf8841c6136a390b0c : SUCCESS in 5m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/8fb6b7a9e13144ada4bd0a5ace8accbb : FAILURE in 57m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d2d30209ef25495399a29aad6f4c2861 : SUCCESS in 1h 10m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2946c0c364fe478181dcf9ed6f3a49ea : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d3feb4946ff4e8a8b94e318b381e428 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2eab87a0a66c483cae0ea209a43a419b : SUCCESS in 10m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/896b186bfa5340a2a6efb9f18dca4ea3 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce1032e8ff44a6eaeaa67e1a5331795 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de279607c0714b8c84b60a95ed88c7a2 : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/26b830a1e673479c9063fb45ff6621eb : SUCCESS in 8m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8bb22b700b8d426f8494162d06a5f4e8 : SUCCESS in 8m 38s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a685504805d24a448eb23bde07ab3bdd : SUCCESS in 5m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ee587ba4c6ba4885bdddfb0997699f66 : SUCCESS in 1h 54m 00s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3c63b4adcdf7445aa19c98b9072d72c4 : SUCCESS in 1h 48m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/92f6bd806202407aad0769ab7ed9bef3 : SUCCESS in 2h 02m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b29cde4a7008424487a59805bbccf5cd : SUCCESS in 1h 45m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50554754fc814add98e44b9e079fcb84 : SUCCESS in 1h 37m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4109558dce2e44e4ae3c83abd6b4a403 : SUCCESS in 1h 41m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9742115b84e59f7d8da8002f0ff5ed7a35980413","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-14 21:03:51.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/11/757911/1/check-not-vote/ext-glance-zuul/b5f9466 : FAILURE in 15m 34s","accounts_in_message":[],"_revision_number":1},{"id":"dd126cf372e004e1ec55c9ffc86b4db3e551b69b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-21 12:46:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f8344159223bea409b24219fbc1fe73c61c03e7e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-21 13:04:14.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/11/757911/1/check-not-vote/ext-glance-zuul/dd6bba4 : FAILURE in 17m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f1c9844e3fe1e9bb92edb62db9904842eb146cf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 15:06:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/44c829afb11a4873b1ca84ee97d9c597 : SUCCESS in 6m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/acaa11bad3d34709a2f8cb64b59d80ae : SUCCESS in 1h 02m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ccf09ab142a440da9e219c17c296d2b5 : RETRY_LIMIT in 18m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d72b9d51537f4f23b93b6a0116bdf9a3 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/063b0c29ba3e4f808cf00568e251f2ee : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebd025a3935a4422a2290f401ce2ae3f : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c84b81f6bdf240d6a3c65a0eff58c592 : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc84e2d6fbb40b7a751c3840cfb5a65 : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c377194962be47ebbb59e42e129a25e6 : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb1aeaba5840437ca65d77876d11b35a : SUCCESS in 11m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5e09b1079d1341e4b14532924dcd2682 : SUCCESS in 8m 58s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2ca227eb745449dbb1c826876fbd1b37 : SUCCESS in 6m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/61251ad2cbd840a78b481118707ad98c : SUCCESS in 1h 01m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8462704a7ed9459a876ede6a6c3884d7 : FAILURE in 1h 09m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a9c4726ddf514063b9dafa20f25b4cec : SUCCESS in 1h 07m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/be37d6acfc394eb8ae1fd2620a2fcf6f : SUCCESS in 1h 13m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e159732ad5245b38c4b35154b52c64b : SUCCESS in 1h 03m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d09a611d9fea4a6890f5efaa09eb2ed4 : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":1},{"id":"033f195c7843c2daae8dd275b12bad8aaec699b5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-04 19:26:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ce25eaf154528f0ae032d87c3551bf70891bbf53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 06:18:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6c2bbbadbff844a1a12b2221bcf97a29 : SUCCESS in 5m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/d598701e14074241a89c8ed02d1a4ca8 : SUCCESS in 1h 12m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d61eb5f2eb394cc0a5957b44e45576e9 : SUCCESS in 1h 07m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb43e750d8b54c50987b419f9593dc96 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e9e4d437ae24d73a3911e089b3cf82e : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6938e17e646426c99ed7794cb8c6fb2 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/290e58236e754017a65054d1ef754149 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7cadd628ee9477c8f0aa9ce9f5dc73c : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/235e7e0d1f6d47d9802e1741e1ca417c : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3475d0164cfa422e95f5f60b9b4b22fc : SUCCESS in 8m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a874356852354d68b9a035827225d9bf : SUCCESS in 7m 04s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/97796e32a6454b8ea92456c7b0408b8a : SUCCESS in 5m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cacb05ea83be49319cb4649e0724e170 : SUCCESS in 1h 12m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/005ddff9ba354af3b196e5ee24407e9c : SUCCESS in 1h 19m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4b822da1ba4442c4b545da82cb5f332d : SUCCESS in 1h 25m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/28ddb3e46ab74f28b6d5c4624e9b3907 : SUCCESS in 1h 26m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c32b4dd25e348cd846c1e8aa8dc49fd : SUCCESS in 1h 31m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d91010197e284d5db2ec95d2dea94244 : TIMED_OUT in 2h 12m 01s","accounts_in_message":[],"_revision_number":1},{"id":"28444af34f778f2439728244f565559d8d7bdc73","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-05 19:21:56.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/757911/757911/1/check-not-vote/ext-glance-zuul/3029c7b : FAILURE in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"125f30df8e70a6754b1b7a18efd4d5e1b7a52c52","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-12-07 14:52:55.000000000","message":"Patch Set 1: Code-Review-1 Review-Priority+1\n\nHi Sean,\n\nThanks for addressing these. My only concern is the migration script changes we have always been very keen to not change unless we absolutely have to. Mainly for not alerting people that something might have been wrong with them and that they would need to address issues in old migrations.\n\nI was looking the patches merged in hacking and I could not find what is actually causing those #noqa tags to fail. Is there a way we could exclude the glance/db/sqlalchemy/migrate_repo as whole instead of changing the scripts? (We really should never have reason to touch those files anyways.)\n\nIf changing the #noqa tags around is really our only option we should add a note of that in the release notes indicating why those scripts has changed and that the change does not affect any deployments.","accounts_in_message":[],"_revision_number":1},{"id":"6c4a3bdb1b79392883c2460172b43729bc1838ee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-07-26 13:39:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c494be456122bb50ea6d8510c7ce1fe5a27d89de","revisions":{"c494be456122bb50ea6d8510c7ce1fe5a27d89de":{"kind":"REWORK","_number":1,"created":"2020-10-13 16:44:36.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/11/757911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/757911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/757911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/757911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/757911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/757911/1"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-10-13 16:43:19.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-10-13 16:43:19.000000000","tz":-300},"subject":"Address new hacking enforcement issues","message":"Address new hacking enforcement issues\n\nNew rule enforcement has been enabled in the latest hacking code. To\navoid breakage later when these changes are released, this addresses a\nfew of the issues that the newer code will error on.\n\nChange-Id: Iacaddd9ae6a9af3d34b5e58d604badb01360fae7\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c494be456122bb50ea6d8510c7ce1fe5a27d89de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c494be456122bb50ea6d8510c7ce1fe5a27d89de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
