)]}'
{"id":"openstack%2Fglance~757999","triplet_id":"openstack%2Fglance~master~Ic15f6843f549f1108435ef199f5d97fcc5147e1b","project":"openstack/glance","branch":"master","topic":"six","hashtags":[],"change_id":"Ic15f6843f549f1108435ef199f5d97fcc5147e1b","subject":"[Part2] Remove six","status":"ABANDONED","created":"2020-10-14 03:45:58.000000000","updated":"2022-02-16 21:23:19.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d0c65b9f5994218dc52ee49172cc845c6837bb64","_number":757999,"virtual_id_number":757999,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-14 23:53:26.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-14 06:35:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 06:35:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 23:53:26.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"29d9cffc985e2f08038f5e29355b3618f07c326a","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-14 03:45:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4ab221da7b388fcba7a85d19c6c8473ff8c03a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 05:18:58.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f5fe1e6b75b34d83a1b4f3fb18801046 : SUCCESS in 1h 01m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/daf16a14a4b54e9698c4dd4cac4f1422 : SUCCESS in 1h 27m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/054d2e3f905b412a92af675a3f33aff7 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a25ff5afc444cd68fdee6cda9900971 : FAILURE in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad724e662f384561836ec73688618e14 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a360073fe7e498a9c14ea0f3d2902b4 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/432e0a2dfeed4f5590cb22b8c1e15f13 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/46f9931888214b24a9e4360ab1b37634 : SUCCESS in 7m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/02c7ea151c9a4fda96f7094f4136c19b : SUCCESS in 6m 48s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2c3b2fe6327e4b07aa102a37ad856e72 : SUCCESS in 4m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/22352248a9b6496ab9445b35835df717 : FAILURE in 31m 50s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/759f588f22914166a602b167521819ca : SUCCESS in 1h 14m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a92f140401fa4307b991cf02ba9d7019 : SUCCESS in 1h 08m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/12ceb54fcd9c41aca42c3559d5c1869e : SUCCESS in 1h 08m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9274a22fca794470b12b4406a97ffe2a : SUCCESS in 1h 03m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a001bca466c94a43b915efd03cafefce : SUCCESS in 1h 18m 39s","accounts_in_message":[],"_revision_number":1},{"id":"3d29718676f45ee98a78d740d6c696f47d70cef6","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-14 05:21:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"47a86549b31f6636779433f52063efecebb4e759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 06:35:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/6e9b906f142f4afcbc86c984673f9961 : SUCCESS in 52m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/722ce19c094f48249b0fea41f0e4ab5e : SUCCESS in 1h 06m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7268061cb9e4577a507b49c770a8d95 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34d27d214ccd41cebf51229b423dfb72 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d3ddcc7b54c48ed81f9ad2327ba062b : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4098e2e8c3f4282a6ee5363ed94c753 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad68f3363e9f4c03bac284ed3971c91a : SUCCESS in 5m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/26042280e0184b7bbedcb9fe089c75c9 : SUCCESS in 7m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/639be9845be94b14b09d1a1af7412cbb : SUCCESS in 9m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2197252f54f846c39d6a1d9e5257fcd2 : SUCCESS in 4m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6c59267e0656475b8be213fbfe5f6889 : SUCCESS in 58m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4c2941fde3854f8580e0533fa985562e : SUCCESS in 1h 12m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9362ddb74bc64958a985bdb07a55d88f : FAILURE in 1h 04m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/41b1878975404b6a845280bb462ab66b : FAILURE in 1h 04m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccafc13b3f5f476c86cd9bf11ee24fe3 : SUCCESS in 59m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ec9fa0ad0d694415b78d457b5639e7fb : SUCCESS in 1h 06m 05s","accounts_in_message":[],"_revision_number":2},{"id":"58c7cd1b7722e7533469fcea7547c426ac8550ea","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-14 23:53:26.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/99/757999/2/check-not-vote/ext-glance-zuul/b3ef93b : FAILURE in 18m 23s","accounts_in_message":[],"_revision_number":2},{"id":"d165fa25ed56ae1a7c954e80ac268adbae983fc5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-05-04 21:20:36.000000000","message":"Topic set to six","accounts_in_message":[],"_revision_number":2},{"id":"d0c65b9f5994218dc52ee49172cc845c6837bb64","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-16 21:23:19.000000000","message":"Abandoned\n\nWe removed six in another series of commits by Stephen Finucane.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5387f2e63a7a8c4f2a01dd5e4aade475f742ad1b","revisions":{"411b57879af839ddd459e42b0aec53570ae268c6":{"kind":"REWORK","_number":1,"created":"2020-10-14 03:45:58.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/99/757999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/757999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/757999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/757999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/757999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/757999/1"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-14 03:40:57.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-14 03:43:23.000000000","tz":480},"subject":"[Part2] Remove six","message":"[Part2] Remove six\n\nWe don\u0027t need this in a Python 3-only world.\nRemove six in follows:\n1. glance/api/v2/image_tags.py\n2. glance/api/v2/images.py\n3. glance/api/v2/metadef_namespaces.py\n\nChange-Id: Ic15f6843f549f1108435ef199f5d97fcc5147e1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/411b57879af839ddd459e42b0aec53570ae268c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/411b57879af839ddd459e42b0aec53570ae268c6"}]},"branch":"refs/heads/master"},"5387f2e63a7a8c4f2a01dd5e4aade475f742ad1b":{"kind":"REWORK","_number":2,"created":"2020-10-14 05:21:49.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/99/757999/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/757999/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/757999/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/757999/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/757999/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/757999/2"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-14 03:40:57.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-14 05:20:47.000000000","tz":480},"subject":"[Part2] Remove six","message":"[Part2] Remove six\n\nWe don\u0027t need this in a Python 3-only world.\nRemove six in follows:\n1. glance/api/v2/image_tags.py\n2. glance/api/v2/images.py\n3. glance/api/v2/metadef_namespaces.py\n\nChange-Id: Ic15f6843f549f1108435ef199f5d97fcc5147e1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5387f2e63a7a8c4f2a01dd5e4aade475f742ad1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5387f2e63a7a8c4f2a01dd5e4aade475f742ad1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
