)]}'
{"id":"openstack%2Fglance~758015","triplet_id":"openstack%2Fglance~master~Ibb0514b58d36f359f5ed2553fbb7673a03bf1151","project":"openstack/glance","branch":"master","topic":"six","hashtags":[],"change_id":"Ibb0514b58d36f359f5ed2553fbb7673a03bf1151","subject":"[Part23] Remove six","status":"ABANDONED","created":"2020-10-14 07:27:59.000000000","updated":"2022-02-16 21:22:36.000000000","total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6a67c665faec6b1046bd53af991838b2fe006fb4","_number":758015,"virtual_id_number":758015,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-14 20:59:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 20:59:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa7326a2e19abc192b3161aa83e2ad79cb39aafb","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-14 07:27:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cfee952fe007a975431aded33f887f942c5ecb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 12:40:44.000000000","message":"Patch Set 1: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfe1e0f105e54330963f110eb4b3ff5d : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d531d3a3b4d84d0d847cc2e0cfb36a9d : FAILURE in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/780e10dba8ad4a60b02262a222801938 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/074910e8e7924258a89ddc4a0864f2fc : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c460ee64fb94c3d979ea0d62601abc8 : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0760caa252604e41bd4707fcd0abdd1b : FAILURE in 7m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/473f861378d04a898a5c11f5b19edc9a : FAILURE in 8m 13s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a8a04830ea514f6c9ccaf6a999c390e5 : SUCCESS in 4m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1ec6768575434329b8bb8f651852ad42 : SUCCESS in 1h 23m 13s","accounts_in_message":[],"_revision_number":1},{"id":"05ad7905681a1439264edc9a34aaefdba0b5e164","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-10-14 13:38:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5ca4b2053975f645198f55ca2b824f7d88a13c10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 20:59:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5607b5138ed4582a20952d3540fb595 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/668e2e5be32e4e56996ea00a7ee94e55 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af3dc2953b5c417384ae70122d0c1d12 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cd9e5b8eb4f4fa480a0e5f662aadc80 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fc13d98b35146c9b333ca29c608bbcc : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/94565d8b38c64d59bed4a482e0749520 : SUCCESS in 8m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9a86f3312b8a42baac6ea485315c2357 : SUCCESS in 8m 59s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/dd3ec339824f4775aeafe83821ea761d : SUCCESS in 4m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a3e7d645cac44fa3b99a566e574f17de : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":2},{"id":"a640aca91a23f84e69e3e109cbaa1715125185e6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-05-04 21:21:09.000000000","message":"Topic set to six","accounts_in_message":[],"_revision_number":2},{"id":"6a67c665faec6b1046bd53af991838b2fe006fb4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-02-16 21:22:36.000000000","message":"Abandoned\n\nWe removed six in another series of commits by Stephen Finucane.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7697f06cb5babbda403d94b6e060472de8627829","revisions":{"c36273a331f2335a59d72024beeacc416ca52a17":{"kind":"REWORK","_number":1,"created":"2020-10-14 07:27:59.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/15/758015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/758015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/758015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/758015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/758015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/758015/1"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-14 07:25:24.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-14 07:27:35.000000000","tz":480},"subject":"[Part23] Remove six","message":"[Part23] Remove six\n\nWe don\u0027t need this in a Python 3-only world.\nRemove six in follows:\n1. glance/tests/functional/test_api.py\n2. glance/tests/functional/test_cache_middleware.py\n3. glance/tests/functional/test_client_exceptions.py\n4. glance/tests/functional/test_client_redirects.py\n5. glance/tests/functional/test_cors_middleware.py\n6. glance/tests/functional/test_healthcheck_middleware.py\n7. glance/tests/functional/test_logging.py\n8. glance/tests/functional/test_reload.py\n\nChange-Id: Ibb0514b58d36f359f5ed2553fbb7673a03bf1151\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c36273a331f2335a59d72024beeacc416ca52a17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c36273a331f2335a59d72024beeacc416ca52a17"}]},"branch":"refs/heads/master"},"7697f06cb5babbda403d94b6e060472de8627829":{"kind":"REWORK","_number":2,"created":"2020-10-14 13:38:16.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/15/758015/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/758015/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/758015/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/758015/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/758015/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/758015/2"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-14 07:25:24.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-10-14 13:37:52.000000000","tz":480},"subject":"[Part23] Remove six","message":"[Part23] Remove six\n\nWe don\u0027t need this in a Python 3-only world.\nRemove six in follows:\n1. glance/tests/functional/test_api.py\n2. glance/tests/functional/test_cache_middleware.py\n3. glance/tests/functional/test_client_exceptions.py\n4. glance/tests/functional/test_client_redirects.py\n5. glance/tests/functional/test_cors_middleware.py\n6. glance/tests/functional/test_healthcheck_middleware.py\n7. glance/tests/functional/test_logging.py\n8. glance/tests/functional/test_reload.py\n\nChange-Id: Ibb0514b58d36f359f5ed2553fbb7673a03bf1151\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7697f06cb5babbda403d94b6e060472de8627829"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7697f06cb5babbda403d94b6e060472de8627829"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
