)]}'
{"id":"openstack%2Fglance~760289","triplet_id":"openstack%2Fglance~master~Ic938cbd22f91000d284b00e3d490f1fb1f1ffba6","project":"openstack/glance","branch":"master","topic":"mox3","hashtags":[],"change_id":"Ic938cbd22f91000d284b00e3d490f1fb1f1ffba6","subject":"remove mox3","status":"ABANDONED","created":"2020-10-29 07:19:42.000000000","updated":"2022-04-13 14:11:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4d4415c915afa083977897329d721ecac17ed751","_number":760289,"virtual_id_number":760289,"owner":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-30 17:00:10.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-30 01:38:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-29 14:55:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-30 02:09:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-29 14:55:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-30 01:38:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-30 02:09:43.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-10-30 17:00:10.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3fa0a95adf366550d7275396cf89b6c75d7f464a","author":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"date":"2020-10-29 07:19:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c57899b0137cd4e9582845e98988bab921d25e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 07:19:54.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"5ad4e4d8d57d0194d137c4783d93017c3db48e61","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-29 07:20:07.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"322ef59b5539cf1ab5bd0264937c3e4201979c8a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-29 14:53:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3a4a72dd46628ae75b1da2227a563a5fedc737c3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-29 14:55:10.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks like this dependency isn\u0027t being pulled in indirectly anymore, so should be safe to remove.\n\n$ tox -re venv --notest\n[clip]\n$ .tox/venv/bin/pip freeze | grep mox3\n$","accounts_in_message":[],"_revision_number":2},{"id":"bfa870409730d3f1d97f0dda4afd238d42521672","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 01:38:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f18271add6e2443cae401b697fb5d11a : SUCCESS in 2m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aeea1714b96c4b7289aba5b646475557 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a39043ac37ab4dce8c67e038e0afb3e9 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdc984f72c92427d8cdc627c368e7a36 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be75dbe17d8745ea9129f3b4caed1a70 : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b01b1b9f4794d07a7aa52e484f7b143 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3f88194034fe45138871e8a48b8f745a : SUCCESS in 8m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3c122abc9c7a41cea45977a1a7f95f72 : SUCCESS in 8m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f9b3ce775d00492aa9a5a63b5850ac14 : SUCCESS in 1h 18m 47s","accounts_in_message":[],"_revision_number":2},{"id":"7372592932cc71e2f359b3ea12d532010151f19b","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-30 02:09:43.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5b11e9607b16ffa1e4878e3bb9de12ff28a689f7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-30 17:00:10.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/89/760289/2/check-not-vote/ext-glance-zuul/89920f6 : FAILURE in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"4d4415c915afa083977897329d721ecac17ed751","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-04-13 14:11:59.000000000","message":"Abandoned\n\nWe no longer use lower-constraints.txt","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a8949745a42d3761dc58dda7576d1577d2b40596","revisions":{"988410f105e3a900f5ad0888d28f5fb131116aa9":{"kind":"REWORK","_number":1,"created":"2020-10-29 07:19:42.000000000","uploader":{"_account_id":28654,"name":"chenming","email":"chenm30@chinaunicom.cn","username":"mmbbz"},"ref":"refs/changes/89/760289/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/89/760289/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/760289/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/760289/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/760289/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/89/760289/1"}}},"commit":{"parents":[{"commit":"b8a34d31284227d29c467ffe33de1873e21a4633","subject":"Merge \"Don\u0027t target broken ssl tests to specific py3.minor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8a34d31284227d29c467ffe33de1873e21a4633"}]}],"author":{"name":"chenming","email":"chenm30@chinaunicom.cn","date":"2020-10-29 07:13:08.000000000","tz":480},"committer":{"name":"chenming","email":"chenm30@chinaunicom.cn","date":"2020-10-29 07:13:08.000000000","tz":480},"subject":"remove mox3","message":"remove mox3\n\nChange-Id: Ic938cbd22f91000d284b00e3d490f1fb1f1ffba6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/988410f105e3a900f5ad0888d28f5fb131116aa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/988410f105e3a900f5ad0888d28f5fb131116aa9"}]},"branch":"refs/heads/master"},"a8949745a42d3761dc58dda7576d1577d2b40596":{"kind":"REWORK","_number":2,"created":"2020-10-29 14:53:23.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/89/760289/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/89/760289/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/760289/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/760289/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/89/760289/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/89/760289/2"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"chenming","email":"chenm30@chinaunicom.cn","date":"2020-10-29 07:13:08.000000000","tz":480},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-10-29 14:53:15.000000000","tz":-300},"subject":"remove mox3","message":"remove mox3\n\nChange-Id: Ic938cbd22f91000d284b00e3d490f1fb1f1ffba6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a8949745a42d3761dc58dda7576d1577d2b40596"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a8949745a42d3761dc58dda7576d1577d2b40596"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
