)]}'
{"id":"openstack%2Fglance~760531","triplet_id":"openstack%2Fglance~master~Ic71710aebafa38dceeead104b3e97fafa4bb4362","project":"openstack/glance","branch":"master","topic":"bp/remove-sqlalchemy-migrate","hashtags":[],"change_id":"Ic71710aebafa38dceeead104b3e97fafa4bb4362","subject":"db: Replace sqlalchemy-migrate call with alembic equivalent","status":"MERGED","created":"2020-10-30 10:03:30.000000000","updated":"2021-07-22 07:59:18.000000000","submitted":"2021-07-22 07:57:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"760531-bp/remove-sqlalchemy-migrate","meta_rev_id":"eb119db42930e99245333461730d2eb4e74ecdb0","_number":760531,"virtual_id_number":760531,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-22 07:57:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-07-21 04:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-07-22 00:27:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-07-22 00:27:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-12 19:49:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-12 19:52:03.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-19 18:55:55.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-07-19 19:01:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2021-07-19 19:01:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-07-21 04:52:51.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"0e0c3736e8cf5fe34a0378521496100d767acfe0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-10-30 10:03:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1ad325bd5568720aa4a15416ccb6b036a118c0e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-10-30 13:16:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9af258d7ee939d3691f1c2341fd283392bb33b0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 22:14:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/fd0dd922b766482f9e23a635eae15185 : SUCCESS in 1h 09m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd2c8028b74b433aa0b1dbaa95d51d70 : FAILURE in 31m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04b5b16d336a41a3b8ef8d2be2d3ee8f : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f553d11f8d5d4760be9728ebfe456e43 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94d85b847ee54dfca80317310d81dbf2 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb17e0c1bca343ceb333dc7a601f2bbb : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e418d1375bdb49f4b917882c093ca9d4 : SUCCESS in 7m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b6c0fbb02a4c4febb84f81594edf2e0c : FAILURE in 8m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4893ecc22d024f8cb749828a852c6aa1 : FAILURE in 9m 04s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9a5e9f62ae6d4d29b06783ddd14c468c : SUCCESS in 5m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/776ff5964c634c2a9ed48c2a179530d6 : FAILURE in 30m 03s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d2d7463b7fc44092844dedf88953aa97 : POST_FAILURE in 31m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/25f865ed8f8642fc86bdc027f0787936 : POST_FAILURE in 22m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/368d314aa7814e7baf6319515c8ca0ff : POST_FAILURE in 23m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6aee3dada4542b1ac56ebe61e97bc8a : FAILURE in 29m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/951cc3f3589d4d44ad229658d5d67610 : FAILURE in 48m 07s","accounts_in_message":[],"_revision_number":2},{"id":"0dc7f260fc30773509bcd8488581b83682746709","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-10-31 01:04:49.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/31/760531/2/check-not-vote/ext-glance-zuul/b22b4ae : FAILURE in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ea3d769753f32e6c44229a5932007b86095c3e25","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-11-02 11:43:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aff06b5e248151f458be8eceaf71a14dc1d64dbd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-11-02 11:48:57.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/31/760531/3/check-not-vote/ext-glance-zuul/c422b2b : FAILURE in 4m 57s","accounts_in_message":[],"_revision_number":3},{"id":"e238d430505435da2f044e66a4b7fae0d24f57f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 20:21:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/09349f74a8ae42b2a6ed3f7ebed19d39 : SUCCESS in 1h 03m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b36763c4cec4f8ebed0e51f5090145a : FAILURE in 1h 01m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a16b2eb655c464a8ed1b6f44f6c926c : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc32630b0f8b4273a0008423cea6a112 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca7da0a4103a4a2baf24beefa605ac51 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87e0acc1a56645a29452cdb32d015a61 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e0c37f1495e404faf3ee444f46eedd9 : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4c0c1d6ae14646debf4024416bf5919c : SUCCESS in 7m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ada965ee2a248aea813e6d2da4c1947 : SUCCESS in 7m 30s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0854674045c64497a5616df4b3a4af88 : SUCCESS in 5m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/eeb383e473d54575b1e1345efb237b76 : SUCCESS in 1h 07m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ca9cd88b79c345a9a1ea04401731b7d2 : SUCCESS in 1h 14m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/04a7e697f2274c09a6a1e113475041fb : SUCCESS in 1h 14m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/698f1a92aa8c4539b72a6124e63613e6 : SUCCESS in 1h 20m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80e1b75e4f194e6b90aea68fc16ee180 : SUCCESS in 1h 00m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/30fef83545a9492d877728d4c69ec499 : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":3},{"id":"27d812c255d6e2e6f3e44a8033797143357d60ec","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-11-06 17:58:21.000000000","message":"Patch Set 3:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":3},{"id":"c92a5413d019a88a40321321656dbd4e713631c4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-11-06 18:18:46.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/31/760531/3/check-not-vote/ext-glance-zuul/49f7673 : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":3},{"id":"a83648f05f21355944a3641329f1adb956075aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 19:35:04.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/24c3fec8872641d29a798f2ea87fd76a : SUCCESS in 1h 02m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dc15d84299ad4175b3467e50775a8bf5 : FAILURE in 1h 06m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f6143df875a4c73afa672e7597a2e50 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e36b396dc3e439ba0e6142785a04009 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9350a4d205764cdf98cf80b689d2839e : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a80d2bdab4754d71938820bd6cfb06d4 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91494b0d55d9490d813e5be51b89e198 : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/755db1ec40254fb4a5ed2c212beafe3b : SUCCESS in 11m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/89737affcff94ed5ab87be3b4e3489dd : SUCCESS in 8m 36s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/dfa7bfea1b60475a91f814b74a811982 : SUCCESS in 4m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d50b6623c30543199f48b139a3badd3e : SUCCESS in 1h 11m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f5813b7243d74663bdc9a7150b045d2b : FAILURE in 1h 08m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d9de9f3d450744cd9b499d1be333d7e5 : FAILURE in 1h 26m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1e501566140a4dbab1a6ab9e93f73508 : SUCCESS in 1h 11m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bb9a51dcd4d4e37803f9e9c92036d59 : SUCCESS in 1h 01m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/acf352278adb46eb929fbbfeb62efdea : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":3},{"id":"3c86be6bbdeb13da531706db1dc5d99ac20a897f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-11-12 14:06:37.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bc1647e0726018caf45a115e88deb25143053ee1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 19:49:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/868a74f17bf846ecae0839671af6a796 : SUCCESS in 1h 12m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/37aa4e56dcea43868dd0ea28d24e4076 : SUCCESS in 1h 13m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0dd1e9cd86547f8b885c3ca9f33bdc6 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc0d4b81c3645a89120eafa35871114 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afa5edbcb9564877902a1886cfa43fde : SUCCESS in 9m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5737415aa96f47ca8b3b82e63b1ace72 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0279dbc526e477d81e0dbbff533db3f : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dc3a23f70d3b497ea7a832cbf060995b : SUCCESS in 8m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c889bf4b78104304a2c4b76fc909a77f : SUCCESS in 8m 00s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/81ec51dbf097493c8e963c111ee78257 : SUCCESS in 6m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f56432a1ea544ecd97bfa78e9f42fe08 : FAILURE in 53m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0f00668a4c6944fbae7b60ce4a930e43 : SUCCESS in 1h 28m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7ea07f1a34794c3b8a03ecca6bcc3073 : SUCCESS in 1h 13m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b20916d65a3f42a89ebbbbef4224f6cc : SUCCESS in 1h 30m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5f97af0a9ea49efa154facfb263e7cf : SUCCESS in 55m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7a58a7b1d0114529a6cbc6165eab47ed : SUCCESS in 1h 19m 29s","accounts_in_message":[],"_revision_number":3},{"id":"2b8c563b2885ef6c0f8889b64477aa5676ae2f0a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-11-12 19:52:03.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/31/760531/3/check-not-vote/ext-glance-zuul/9fa7254 : FAILURE in 4m 32s","accounts_in_message":[],"_revision_number":3},{"id":"e6c4e8ab58a13d2e44825cb1f0261a49fc93331e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-13 16:19:46.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"02d79633dcfe5a5e1c5b0c65dc7e43035f09afda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 18:04:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e058985e00e3418fbadd669b6af8b29f : SUCCESS in 1h 01m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f6a9d39578844efdabe4680aecfcf292 : SUCCESS in 1h 23m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d17f6291c844f7b857f46b3020fc34e : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7e97d51cf3a486f9bea486fe1fb2374 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1f5342bbfc643ab91258a8dafc01252 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac383b501ef94e7aaf02d8062e98dcc3 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7b4cbd0c22a47c68fb5cf1a005b2032 : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b123e232a2841b0ac0835d35df90c29 : SUCCESS in 8m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2531abab392f4cf5a7001175de650dd2 : SUCCESS in 8m 23s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/d205f04cecd043ec96b4f890c931a67f : FAILURE in 7m 38s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c7a9f12acfc74ce59c2266ca3f67e546 : SUCCESS in 6m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5f716a68aea940d3866e7d2591dd37bc : SUCCESS in 1h 06m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e2c4a06fac9144eab9e2c16e6386a967 : SUCCESS in 1h 37m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e0f7fc2090224d9eaf6fb3bdfe25c721 : SUCCESS in 59m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8e7eefa6637640fd93b478e8b92195c2 : SUCCESS in 1h 17m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9eb591d912234ba1a8d3f8ffd40191a8 : SUCCESS in 43m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8d3c53df6e974d67b4dac48058ff8dfb : SUCCESS in 1h 28m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cc07f8061147493cb32e1cc3dfb6bc03 : SUCCESS in 32m 23s","accounts_in_message":[],"_revision_number":4},{"id":"9b9fc78a7797bd0e85750e695e7093cdd15b8c13","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-19 18:55:55.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nOne question inline","accounts_in_message":[],"_revision_number":4},{"id":"51b94712d16c14513619e07bb8496438382bea13","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-20 09:03:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b9e8e8ecbe0f111d3089b4190f3711cc0c414653","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-21 04:52:51.000000000","message":"Patch Set 4: Code-Review+2\n\nThank you Stephen!!","accounts_in_message":[],"_revision_number":4},{"id":"a464f6a74efb3b8b015e41f8a475ef926ca0c19d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-07-22 00:27:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3d74af43a8163dd279b2edd6d6dcfaf7fec855a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 00:28:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9e1170bbad66be34e0b82878ca700e73b8dc235a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 03:45:17.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/104dfac104354cd495987f21d2077440 : SUCCESS in 53m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a1ec4e4a7657490c8adae3271d042941 : SUCCESS in 1h 00m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24a601ee60754564a532dcd4bce412ac : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ce7f84111ab462eadb4f6c76ebca477 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf88df64e5904bc5a6dc7b4b8f676105 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4a149607bb14714aa5feb1ebcbfcb93 : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f13a3b6e860f497a9fc80b5999b90eca : TIMED_OUT in 31m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/27f6e914dcb24333ba8497162cfe8766 : SUCCESS in 8m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7fb1537173d2427cbc32ba6f35653e53 : SUCCESS in 55m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a69cbcb57ba44d9aa4eec6357c8cef2f : SUCCESS in 1h 06m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee1044e20bee4d4b82e12005385e3333 : SUCCESS in 39m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f0b2cfe814954d30afb0b22dd184f106 : SUCCESS in 1h 19m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9a30b5f252154fc68dd1792e1dbfa25a : SUCCESS in 28m 51s","accounts_in_message":[],"_revision_number":4},{"id":"01aa9e0742466ec6b96557da44db2c4fcb2ff8bf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-22 05:04:37.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9eae0af8dc922773ae0d09478c8d4074197360e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 06:35:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/71bad124822e461783f051436f4f31cd : SUCCESS in 1h 06m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c55db2425964c04a8e0bcc7eeaf6b65 : SUCCESS in 1h 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a06540d2fbdf44939bc791cebf9a1d28 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/806f754acf8346d1b3dcb449ee9b55a0 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b743ac1969b244d9ae4636afb86809bc : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f609403346e4ce2b31d6285c17c29d2 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e88862bd32ff45db89c9bd1c397c9dac : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a304eaf3b17405c96a1d17438f9e6ca : SUCCESS in 8m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b8b31b6df14a48a3b37a2fca01343337 : SUCCESS in 9m 00s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/2049b69e81ef410a9e9a3cfcfa33d7a3 : SUCCESS in 8m 07s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/44a8d0b50b1b437ab4897aad8cae33fe : SUCCESS in 4m 51s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a660bed8988845a1b99b35eb1e991106 : SUCCESS in 1h 15m 36s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/29ca5f1b64734453a7d8b6ef62967547 : SUCCESS in 1h 23m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/29264f8670b74c3f84c2e88915f5f0c2 : SUCCESS in 1h 15m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0a71e0edc17646eebe17cd75eca9870d : SUCCESS in 1h 22m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/149a7e7ab415414fbbb21f612286c432 : SUCCESS in 50m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a288105d0c6a426f858bdbcfc016ae8a : SUCCESS in 1h 19m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a14313803de04119ba1df1543be72625 : SUCCESS in 34m 00s","accounts_in_message":[],"_revision_number":4},{"id":"e11018507eee4612886fcd2381a7d8ae34d50358","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 06:36:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f6f116e1fc69224eb1925f41cd6361b0c5064ab0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 07:57:23.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9390a7461d704532b5f9646bd6b8bf1b : SUCCESS in 1h 04m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b57497b1177441cebd8f48352ee89703 : SUCCESS in 1h 09m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f11e50444a46b882f656ab2c6dfe99 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce67ed11132743059af78e1584ac589b : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57b5cb6c7c0043b68dc55dd2288f5c3b : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb16c63e001f4fbbb00947a2294c20b3 : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a2d914cb8e714d32a646ec88b1a28b45 : SUCCESS in 8m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/849ad0dc34264e1c868f71b421463332 : SUCCESS in 9m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2c6c1d6fed1c4474a1ac5a223425c239 : SUCCESS in 1h 05m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8d6054980735424a9958d3bba05bba1d : SUCCESS in 51m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88ed32320abd41cd8d0a86f6d2f516c8 : SUCCESS in 51m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/68638e20c4d5477d8c4869ca2708ebf2 : SUCCESS in 1h 15m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e29c546609c0455ab3391b83757064a4 : SUCCESS in 35m 30s","accounts_in_message":[],"_revision_number":4},{"id":"e9cb2d7f25984ab5331486c13ebd694fb6cd0cc1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 07:57:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"eb119db42930e99245333461730d2eb4e74ecdb0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 07:59:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e132ea99171b4da7a7eb0404fd266bbb : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"77676e833f861373e07de7ab04c6a56c8d37fc78","revisions":{"f2c83819ae97cf0a02e7b2b5ad4c77789ba79faf":{"kind":"REWORK","_number":1,"created":"2020-10-30 10:03:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/760531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/760531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/760531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/760531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/760531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/760531/1"}}},"commit":{"parents":[{"commit":"7143eb8c18133f9ca1bb9a5d0cf4ec2e7d84c25a","subject":"db: Move \u0027schema\u0027 module up a level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7143eb8c18133f9ca1bb9a5d0cf4ec2e7d84c25a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-10-30 09:47:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-10-30 09:49:04.000000000","tz":0},"subject":"db: Replace sqlalchemy-migrate call with alembic equivalent","message":"db: Replace sqlalchemy-migrate call with alembic equivalent\n\nThe \u0027ocata_expand01_add_visibility\u0027 alembic migration was calling\n\u0027alter\u0027 on a table column, something that is only possible once you\u0027ve\nmonkey patched sqlalchemy by importing the \u0027migrate\u0027 module. Resolve\nthis.\n\nChange-Id: Ic71710aebafa38dceeead104b3e97fafa4bb4362\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2c83819ae97cf0a02e7b2b5ad4c77789ba79faf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2c83819ae97cf0a02e7b2b5ad4c77789ba79faf"}]},"branch":"refs/heads/master"},"16a5b17d75dd13ceb216c63f3c46587ccabb1051":{"kind":"REWORK","_number":2,"created":"2020-10-30 13:16:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/760531/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/760531/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/760531/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/760531/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/760531/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/760531/2"}}},"commit":{"parents":[{"commit":"7143eb8c18133f9ca1bb9a5d0cf4ec2e7d84c25a","subject":"db: Move \u0027schema\u0027 module up a level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7143eb8c18133f9ca1bb9a5d0cf4ec2e7d84c25a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-10-30 09:47:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-10-30 13:12:34.000000000","tz":0},"subject":"db: Replace sqlalchemy-migrate call with alembic equivalent","message":"db: Replace sqlalchemy-migrate call with alembic equivalent\n\nThe \u0027ocata_expand01_add_visibility\u0027 alembic migration was calling\n\u0027alter\u0027 on a table column, something that is only possible once you\u0027ve\nmonkey patched sqlalchemy by importing the \u0027migrate\u0027 module. Resolve\nthis.\n\nChange-Id: Ic71710aebafa38dceeead104b3e97fafa4bb4362\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/16a5b17d75dd13ceb216c63f3c46587ccabb1051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/16a5b17d75dd13ceb216c63f3c46587ccabb1051"}]},"branch":"refs/heads/master"},"8c9616f7f0c3a9b1589ec50065039412af0d950c":{"kind":"REWORK","_number":3,"created":"2020-11-02 11:43:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/760531/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/760531/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/760531/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/760531/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/760531/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/760531/3"}}},"commit":{"parents":[{"commit":"7143eb8c18133f9ca1bb9a5d0cf4ec2e7d84c25a","subject":"db: Move \u0027schema\u0027 module up a level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7143eb8c18133f9ca1bb9a5d0cf4ec2e7d84c25a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-10-30 09:47:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-11-02 11:43:22.000000000","tz":0},"subject":"db: Replace sqlalchemy-migrate call with alembic equivalent","message":"db: Replace sqlalchemy-migrate call with alembic equivalent\n\nThe \u0027ocata_expand01_add_visibility\u0027 alembic migration was calling\n\u0027alter\u0027 on a table column, something that is only possible once you\u0027ve\nmonkey patched sqlalchemy by importing the \u0027migrate\u0027 module. Resolve\nthis.\n\nChange-Id: Ic71710aebafa38dceeead104b3e97fafa4bb4362\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c9616f7f0c3a9b1589ec50065039412af0d950c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c9616f7f0c3a9b1589ec50065039412af0d950c"}]},"branch":"refs/heads/master"},"77676e833f861373e07de7ab04c6a56c8d37fc78":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-13 16:19:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/760531/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/760531/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/760531/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/760531/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/760531/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/760531/4"}}},"commit":{"parents":[{"commit":"a75aa3863ac238dce2cb95018bb058f71583347e","subject":"db: Move \u0027schema\u0027 module up a level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a75aa3863ac238dce2cb95018bb058f71583347e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-10-30 09:47:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-07-13 16:15:41.000000000","tz":60},"subject":"db: Replace sqlalchemy-migrate call with alembic equivalent","message":"db: Replace sqlalchemy-migrate call with alembic equivalent\n\nThe \u0027ocata_expand01_add_visibility\u0027 alembic migration was calling\n\u0027alter\u0027 on a table column, something that is only possible once you\u0027ve\nmonkey patched sqlalchemy by importing the \u0027migrate\u0027 module. Resolve\nthis.\n\nChange-Id: Ic71710aebafa38dceeead104b3e97fafa4bb4362\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77676e833f861373e07de7ab04c6a56c8d37fc78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77676e833f861373e07de7ab04c6a56c8d37fc78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
