)]}'
{"id":"openstack%2Fglance~761938","triplet_id":"openstack%2Fglance~stable%2Fqueens~Ie98c4287c8bd1c364655adb99c7a88d1d451496e","project":"openstack/glance","branch":"stable/queens","topic":"bug/1800601","hashtags":[],"change_id":"Ie98c4287c8bd1c364655adb99c7a88d1d451496e","subject":"py3: fix recursion issue under py37","status":"ABANDONED","created":"2020-11-09 14:39:06.000000000","updated":"2022-05-11 14:29:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5d4768081e48b8d391e3e28615f244b535cf665a","_number":761938,"virtual_id_number":761938,"owner":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-09 20:33:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"all":[{"value":2,"date":"2020-11-09 14:54:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":1,"date":"2020-11-26 09:41:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-11-09 14:39:06.000000000","_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"}],"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-09 14:39:06.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"},{"updated":"2020-11-09 14:54:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-11-09 20:33:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-26 09:41:21.000000000","updated_by":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"reviewer":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"state":"REVIEWER"},{"updated":"2020-12-14 17:41:54.000000000","updated_by":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"reviewer":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"state":"CC"}],"messages":[{"id":"6802c223efc7267b876109e8582a008b0f0bd20f","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2020-11-09 14:39:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d65b84a2d685e99b091601925427ad95860d7bc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-11-09 14:54:47.000000000","message":"Patch Set 1: Code-Review+2\n\n3.7 wasn\u0027t supported in this release, but safe enough.","accounts_in_message":[],"_revision_number":1},{"id":"b4323e3d9f67cb380ab853ee624081c91fea727c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 20:33:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/93becf5cc260424493c58ae0fcf2cdf6 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb042393b1f240d389f841d85cfaf6df : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc866bc313b84d23b456add3ec046826 : SUCCESS in 6m 30s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2cf06a42280f40389c5d45e301c835e0 : SUCCESS in 6m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/18d14b82bbcd43e798be75b1560857fd : SUCCESS in 1h 32m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1b71ad19135f438e9b818b24a07f825d : SUCCESS in 1h 00m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1636ebb6edf4c23b44425186c50d482 : SUCCESS in 1h 24m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21d6c6c9bef24a63b880749cfb9f6192 : SUCCESS in 6m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7f5ba538250a4489837b1f1de42aacad : FAILURE in 9m 05s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/2ba63b2e4ea948b1876bc34e0c71671a : FAILURE in 5m 40s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c27e29f10b4d4f9f9990f9a836b29d44 : FAILURE in 1h 03m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f74f7663434246f2963a6485653ad82bda2be3e6","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-11-26 09:41:21.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"b7b133e0be36dc0921524010b139becd4d346a4e","author":{"_account_id":30998,"name":"Andre Luiz Beltrami Rocha","display_name":"Andre Beltrami","email":"debeltrami@gmail.com","username":"andrebeltrami"},"date":"2020-12-14 17:41:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d11cd4375be35d14caf3baf4adeca81ea817838c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 20:12:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/928743c5f9a9416d84f4d8791e640653 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47450e8565ea47f688976669cf8d39ec : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a60b77f9b6f44237b9c6b3b009e1c2b9 : SUCCESS in 5m 43s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/97e97c43946b4207b668fd828dd23e16 : SUCCESS in 7m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6d46091448e944b2b1106b422dbbf634 : SUCCESS in 1h 12m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/df2188cf43e44f51b8a24381ac385e32 : SUCCESS in 43m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4249c43e7caa4c918c2908ce30f56580 : SUCCESS in 1h 29m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf5b7b14a26b402599b46f68a9a18a03 : SUCCESS in 5m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/aed7d088b48b4f3fbe1aa02c5d7e83b3 : FAILURE in 11m 08s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/421f2b171b8e42f8b2b6eb20e7c32556 : FAILURE in 4m 45s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d6a792d29ef64fc3aebca815949a19ab : FAILURE in 1h 00m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d4768081e48b8d391e3e28615f244b535cf665a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-05-11 14:29:53.000000000","message":"Abandoned\n\nThis does not fit with our backport criteria [0]\n\n[0] https://docs.openstack.org/project-team-guide/stable-branches.html#appropriate-fixes","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7624be1e4cbf7f02f6b4e1c418cb3d0300e45d1c","revisions":{"7624be1e4cbf7f02f6b4e1c418cb3d0300e45d1c":{"kind":"REWORK","_number":1,"created":"2020-11-09 14:39:06.000000000","uploader":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"ref":"refs/changes/38/761938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/761938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/761938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/761938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/761938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/761938/1"}}},"commit":{"parents":[{"commit":"cc4f74d112bb93cec0e51a37eea1d509b5ad1776","subject":"Merge \"Mitigate OSSN-0075\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc4f74d112bb93cec0e51a37eea1d509b5ad1776"}]}],"author":{"name":"James Page","email":"james.page@ubuntu.com","date":"2018-11-02 10:10:49.000000000","tz":0},"committer":{"name":"Lucio Seki","email":"luciomitsuru.seki@fit-tecnologia.org.br","date":"2020-11-09 14:39:00.000000000","tz":-180},"subject":"py3: fix recursion issue under py37","message":"py3: fix recursion issue under py37\n\nAdd tox target and resolve issue with infinite recursion in\nExtraProperties class by directly using methods on the dict\nclass.\n\nCloses-Bug: 1800601\n\nChange-Id: Ie98c4287c8bd1c364655adb99c7a88d1d451496e\n(cherry picked from commit 43587fb93d8d949d9a9895153ac06ea4e7e6d150)\n(cherry picked from commit 2bcd6c66629a06afdfcd32ff7892c5571bbe0573)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7624be1e4cbf7f02f6b4e1c418cb3d0300e45d1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7624be1e4cbf7f02f6b4e1c418cb3d0300e45d1c"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
