)]}'
{"id":"openstack%2Fglance~764107","triplet_id":"openstack%2Fglance~master~I78ec79f97bfdcc39772448296140f8d8f35adea1","project":"openstack/glance","branch":"master","topic":"remove_admin_role_opt","hashtags":[],"change_id":"I78ec79f97bfdcc39772448296140f8d8f35adea1","subject":"Remove \u0027admin_role\u0027 option","status":"MERGED","created":"2020-11-25 01:35:47.000000000","updated":"2020-12-02 22:57:39.000000000","submitted":"2020-12-02 22:55:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764107-remove_admin_role_opt","meta_rev_id":"ee18fa6d57da51088af282b1b5c398558e2a418b","_number":764107,"virtual_id_number":764107,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-02 22:55:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-12-02 18:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-12-02 15:46:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2020-11-26 05:28:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2020-12-01 05:12:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-12-02 18:48:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-25 03:14:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-26 05:28:39.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-11-26 11:21:52.000000000","updated_by":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"reviewer":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"state":"CC"},{"updated":"2020-12-01 05:12:15.000000000","updated_by":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"reviewer":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"state":"REVIEWER"},{"updated":"2020-12-02 15:46:05.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2020-12-02 18:48:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"c709eb26863ab3f1345c028d6e78ccc307610b6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-11-25 01:35:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef837df957d85ecf064a619e911a3debe14dbc1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 03:14:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e0b33362c6ea40f59f2e1fe96f99d6b6 : SUCCESS in 1h 13m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/52acd1beb02f4428bcf9700e353edfe5 : FAILURE in 1h 30m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7345961ac5ab411fbbe75342e9fcae54 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff699d547fa6465eb23fab47e62e6760 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aec08dc80daa43f9b0392eced07799f3 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16e5d51489e3403da5f4894fb85cc59f : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/733d8e5722194bbfa0ea41f1b9d5e3cc : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e6c88670430427fb55bb2dae10139d2 : SUCCESS in 6m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e60cbd47fc834ab79f2553b9592fbb80 : SUCCESS in 8m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/65e24481982b4dfe8b426a2ad6265fa8 : SUCCESS in 7m 48s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0b6338a83e134d05a60723009610c689 : SUCCESS in 4m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/79eb959a7f254f50990281b0f1b9bb90 : SUCCESS in 1h 14m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d364403e615a4007b02b93de23b8adf4 : FAILURE in 1h 11m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1f2217c01c944558b0bf50960e85c448 : FAILURE in 1h 14m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/52290b3e03ba4d68990125aec144f1ae : FAILURE in 1h 21m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/616c1763aea84f45bd1d90bd5194162d : SUCCESS in 1h 10m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1495a98e82b14dd9917128ee878c5b32 : SUCCESS in 1h 06m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b7e2fe88057f55d6d8641e8670404bfcf6db955d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-11-25 13:44:48.000000000","message":"Patch Set 1:\n\ntempest-integrated-storage failures are being looked into; this patch that merged 5 hours ago skips the problematic test until a fix is figured out (it\u0027s a swift issue): https://review.opendev.org/c/openstack/tempest/+/764061","accounts_in_message":[],"_revision_number":1},{"id":"4dffb248fc0b2548e45c95401f0e6dffebb751f0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-11-25 13:45:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"574a48dc3cbf15d683584156c39eb3085b1bb662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 20:17:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/18ae67fdf7564422b2989bb4e4fb2fae : SUCCESS in 1h 04m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/73cff5fdf1aa49b08bd66b6118fd84b2 : SUCCESS in 1h 15m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb7cf14282de474d9247afe9709b3f8e : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faa4638e0a58438385ed6322d806c6bb : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/218c187819df47ecb3e5f826f546b152 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f77970046f4444968440f4bcd6e9e145 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4c64ed833fa4fd89ccecb2090c43707 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90e74ce6e20c4630b08035039c803af9 : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/31420dc0d415423199d9b3dd09d8c566 : SUCCESS in 7m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6983eee0ff3f4a06905e0149e372abca : SUCCESS in 6m 53s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3feba508ad6a45ff987e4aa4ebdd83ba : SUCCESS in 4m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/da6936c0e75143adbf5f7a23befbe23b : SUCCESS in 1h 11m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1d6c533f51fa4dbca8435c9b1f9b7b49 : SUCCESS in 1h 19m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ffdbf223b85d4b55bb7ccb94ca764258 : SUCCESS in 1h 13m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/460dc22f6f974807875f22943b7a0d95 : SUCCESS in 1h 25m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/149022de9b704dcf979418e6aafefbff : SUCCESS in 1h 05m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/02a0d3be007945babc4c63872e44e582 : SUCCESS in 1h 10m 18s","accounts_in_message":[],"_revision_number":1},{"id":"dbff5ee09ffa6387c52ab6207f3b12495a39bea5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-11-26 05:28:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6cd71ef83b5e5fc9a630be59efa151074c470c00","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-11-26 11:21:52.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nHi Brian Rosmaita,\nPlease see my 2 comments","accounts_in_message":[],"_revision_number":1},{"id":"1d9cae13b220a2c1b029de29fb20d1f7c8deb5d7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-11-30 17:02:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8dfaf1832366b095b13f201b2f3f88f94dea70f","author":{"_account_id":32171,"name":"Girish Chilukuri","email":"girish.chilukuri@ibm.com","username":"GirishChilukuri"},"date":"2020-12-01 05:12:15.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"e52fbd1655fe48de506b9bb78433b4c09c0d6713","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-02 15:46:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d7bd329bc74ec263c7dbfdf9725ef329715b67e0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-02 18:48:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks good!","accounts_in_message":[],"_revision_number":1},{"id":"7b00131455141a89c62be82ce52c8a193de3f09f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 18:49:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4ef8460d3d202ea9104ced6098fbaf617f148a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 22:55:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3ac49f9112f54947b354a4c536ec9193 : SUCCESS in 1h 05m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/97bcbeacadc049338dde64dfc5894946 : SUCCESS in 1h 15m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdce6de57e2347f2993c54f4d5a37dc8 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c867f8db820b4f7f87e1e9f3068ddc47 : SUCCESS in 9m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71b499a8af5f4e91925be6fafb5fda0b : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/323f15e4b7da42f6a635267717911486 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70093b5e219440e39f68b1f6761a917b : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e4c25e5498d4fdc971916a52e6d4316 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c310eec3f26940788d8901415fe19abc : SUCCESS in 8m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b2924f2ab7bc436694b24a854b00afe9 : SUCCESS in 7m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/830bf8f41d9a4c1cbe7a3dd92fb7ee03 : SUCCESS in 1h 23m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b586b8ec6cd84e14bb01932967736e0c : SUCCESS in 1h 13m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97d809b0a9e94fb4957054dd01189fcc : SUCCESS in 1h 06m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d07b3a478fcf43f9889b3052e0d4a58b : SUCCESS in 1h 03m 39s","accounts_in_message":[],"_revision_number":1},{"id":"24de5ed647cc3c31d9c0d6af4323891ad2e7faba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 22:55:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ee18fa6d57da51088af282b1b5c398558e2a418b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 22:57:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c288a3bf64e417689312755c78fb664 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/184618b0520146d0894f2e955fe67a11 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f102b74a285ad7de90c99ee7f5d6bef6e0d46161","revisions":{"f102b74a285ad7de90c99ee7f5d6bef6e0d46161":{"kind":"REWORK","_number":1,"created":"2020-11-25 01:35:47.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/07/764107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/764107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/764107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/764107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/764107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/764107/1"}}},"commit":{"parents":[{"commit":"1344c45772c1a77107f621632642cf1488e431db","subject":"Adding gate job for glance cinder store","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1344c45772c1a77107f621632642cf1488e431db"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-11-25 00:55:21.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-11-25 01:31:21.000000000","tz":-300},"subject":"Remove \u0027admin_role\u0027 option","message":"Remove \u0027admin_role\u0027 option\n\nThis option was deprecated in the Ussuri release by change\nI0f61f85a0aaa4f68e345fa08fbb6b039d3d32587 and it is now eligible\nfor removal following the standard OpenStack deprecation policy.\n\nChange-Id: I78ec79f97bfdcc39772448296140f8d8f35adea1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f102b74a285ad7de90c99ee7f5d6bef6e0d46161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f102b74a285ad7de90c99ee7f5d6bef6e0d46161"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
