)]}'
{"id":"openstack%2Fglance~764242","triplet_id":"openstack%2Fglance~master~I6d4de5253205d34eb610b95f2d79db67f2699cc6","project":"openstack/glance","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I6d4de5253205d34eb610b95f2d79db67f2699cc6","subject":"Implement secure RBAC for image tags","status":"ABANDONED","created":"2020-11-25 21:02:22.000000000","updated":"2021-02-19 10:05:00.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"00da608d7da11e0e35cf161d62c880a292a3e028","_number":764242,"virtual_id_number":764242,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-30 23:12:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-26 02:25:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-03 05:23:13.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"}],"messages":[{"id":"1ea030508ecf5c9f6d120049193a34a76ffd0a10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-25 21:02:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b9a1c5215742ed18a765f7170977847eed12a2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-25 22:31:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f12610864cef94c0c658b6862870fd2b6cb4d2df","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-25 22:32:44.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":2},{"id":"dc78030a861a1376dec599644805a82bdf7e295a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 02:25:13.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/1dd0da6ccf40458c97aaefabf01cfaf2 : SUCCESS in 1h 07m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f50a7501681a47bcb28d469894dc7ce7 : FAILURE in 1h 17m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac2a8a0121f64084a769eb75ec142782 : FAILURE in 19m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6ec4e3dec0a43169046c43d45895df0 : FAILURE in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f168c3eea9264919ab95ee86a2218cc6 : FAILURE in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f443d0925f204dd48a03017047f31427 : FAILURE in 15m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5291fece6b1242e49d9be07af4bf6720 : FAILURE in 4m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/450b626f86aa4f4bbbf0c161cb366285 : FAILURE in 15m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d00c8272b0094c66820e78be8a85827e : FAILURE in 8m 12s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1d908d49fb884abda51e7b3e02d21905 : SUCCESS in 5m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/96a0567b9fae4521abfa9601d7681691 : FAILURE in 1h 12m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b77cbed31fee425284346dc96ad63d76 : FAILURE in 1h 09m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/72374f2e62a84c13b003d54eae895ab7 : FAILURE in 1h 02m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3b4e0847de0b4f1294347e96e8990712 : FAILURE in 1h 45m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45fcbc2f28e944f68a5b1861c72500d6 : SUCCESS in 1h 08m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/725c547548ec45b5a0d10ce18d5d4d0c : FAILURE in 1h 12m 11s","accounts_in_message":[],"_revision_number":2},{"id":"0279f1852736fed714685db89435bf275f243bee","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-30 21:40:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"73cc4b4af1af2c934c346632a53c6d6cf53ca7ea","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-30 21:40:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3247c5fd04a52001b75abb0b40ae264400d5c6b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-30 21:49:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cb6381681a5d4d59264c3683d994e335a9b6a08f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 23:12:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5a7c9b22ac39409fa5965ce49053dcc6 : FAILURE in 15m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/326432361c16479390b5e03a3327f344 : FAILURE in 1h 14m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21340e72cb55413db66cb37c11d7d156 : FAILURE in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/082583034163443fa60f2f2d3a60e970 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfa76dca845447cdbf0188923e421b86 : FAILURE in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebdc3db6ba7d4a529a747e84b0719d13 : FAILURE in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bea1a94016742b9b7af88abae4c3e07 : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/711f487e2e214c24a92afaaa4c86673a : FAILURE in 7m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6dd8b83242774fc1a853786f8104d153 : FAILURE in 8m 05s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9352f2c9b5d9459e9d7bfddc48ffbd73 : SUCCESS in 4m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/20c1cd242fb446c082e3168d3a9965e3 : FAILURE in 1h 12m 27s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/65808f387f8a464f8c4aa4109cd3a717 : FAILURE in 1h 15m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/179889c4f67b4af2973d961467e9e072 : FAILURE in 1h 13m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/20a45223d7714b17b2dff86c2a25c008 : FAILURE in 1h 15m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75871e32a31e4dab87448d2f6c6bc4ad : SUCCESS in 1h 06m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8d67e7325edd484e92c0411c0a50aa7f : FAILURE in 1h 13m 26s","accounts_in_message":[],"_revision_number":3},{"id":"243a00753a8ae4242e31b109fa4114df4c09d19b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-03 05:23:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"43bd385ba258c460076f6288fc1c0a70c1ad5674","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 06:36:10.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/9c69b8409ac1449483cc19db45e20831 : SUCCESS in 1h 02m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3df4feb585494dd4969f7ca1a9984fc1 : FAILURE in 1h 03m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1696ed2896fe4137a624d4e8ee46d7f4 : FAILURE in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c7d0212e724e4d835a02b8192e4fae : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63c2142a143a4705b40b96f04e8d1c81 : FAILURE in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe532a68e57f400099169c5a17c34e54 : FAILURE in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60d4fffaccf04aec822ef7d7d9cb5fde : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad2168a8b4df4243b4f2ae7001e7ebd9 : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/08a3af1cc42e4978a7775cd091aef90d : FAILURE in 7m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7fa5ee695fc14eb6a743d7724f9adee4 : FAILURE in 6m 32s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e371efddb8894b6c9d6ff9ebdd661028 : SUCCESS in 4m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d38313f6f037499eae67cb60374a6e93 : FAILURE in 1h 00m 34s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/579a0888f3f04c4882f5f984ab70a1c1 : FAILURE in 1h 03m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/388f11f820144a7090d33a4a64588bcf : FAILURE in 1h 00m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c9b61b8855804fd4b5d1bf6730477d35 : FAILURE in 1h 09m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a61acc2e0ed74137b22127e860ded70f : SUCCESS in 56m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c585f79c5a9a441bb8b01397c859a59d : FAILURE in 1h 07m 40s","accounts_in_message":[],"_revision_number":3},{"id":"00da608d7da11e0e35cf161d62c880a292a3e028","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-19 10:05:00.000000000","message":"Abandoned\n\nWe don\u0027t have policies for tags","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"556e7de0d5083cba322f4ba3de3c07f71a876ff2","revisions":{"2123f9d0d1ad6b73badc92afae2f2b6ab38de140":{"kind":"REWORK","_number":1,"created":"2020-11-25 21:02:22.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/42/764242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/764242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/764242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/764242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/764242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/764242/1"}}},"commit":{"parents":[{"commit":"ab9d43185ba4fb372444a52a892d7ed647b81699","subject":"Add basic/common personas to base policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab9d43185ba4fb372444a52a892d7ed647b81699"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-25 20:57:23.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-25 20:57:23.000000000","tz":0},"subject":"Implement secure RBAC for image tags","message":"Implement secure RBAC for image tags\n\nThis commit adds two new policies for image tags. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC experience and\nimprove security.\n\nChange-Id: I6d4de5253205d34eb610b95f2d79db67f2699cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2123f9d0d1ad6b73badc92afae2f2b6ab38de140"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2123f9d0d1ad6b73badc92afae2f2b6ab38de140"}]},"branch":"refs/heads/master"},"8e88673bc4ea1d7fad29f25cddd42d660e159656":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-25 22:31:48.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/42/764242/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/764242/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/764242/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/764242/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/764242/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/764242/2"}}},"commit":{"parents":[{"commit":"de48694d4f25bca28816fea958ace12156d83417","subject":"Add basic/common personas to base policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de48694d4f25bca28816fea958ace12156d83417"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-25 20:57:23.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-25 21:25:16.000000000","tz":0},"subject":"Implement secure RBAC for image tags","message":"Implement secure RBAC for image tags\n\nThis commit adds two new policies for image tags. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC experience and\nimprove security.\n\nChange-Id: I6d4de5253205d34eb610b95f2d79db67f2699cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e88673bc4ea1d7fad29f25cddd42d660e159656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e88673bc4ea1d7fad29f25cddd42d660e159656"}]},"branch":"refs/heads/master"},"556e7de0d5083cba322f4ba3de3c07f71a876ff2":{"kind":"REWORK","_number":3,"created":"2020-11-30 21:49:03.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/42/764242/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/764242/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/764242/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/764242/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/764242/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/764242/3"}}},"commit":{"parents":[{"commit":"bc2d16581333cb95e0dabeeac55ccd1006c3fb9c","subject":"Add basic/common personas to base policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc2d16581333cb95e0dabeeac55ccd1006c3fb9c"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-25 20:57:23.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-30 21:36:12.000000000","tz":0},"subject":"Implement secure RBAC for image tags","message":"Implement secure RBAC for image tags\n\nThis commit adds two new policies for image tags. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC experience and\nimprove security.\n\nChange-Id: I6d4de5253205d34eb610b95f2d79db67f2699cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/556e7de0d5083cba322f4ba3de3c07f71a876ff2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/556e7de0d5083cba322f4ba3de3c07f71a876ff2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
