)]}'
{"id":"openstack%2Fglance~764251","triplet_id":"openstack%2Fglance~master~I33809f5fde7dd60b27156866744d2a06facd416d","project":"openstack/glance","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I33809f5fde7dd60b27156866744d2a06facd416d","subject":"Implement secure RBAC for metadef APIs","status":"ABANDONED","created":"2020-11-25 22:31:48.000000000","updated":"2021-09-02 18:45:48.000000000","total_comment_count":3,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"d786397f083617924a92e04abe3dd95f8a3b3f39","_number":764251,"virtual_id_number":764251,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-25 06:42:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":-1,"date":"2021-02-26 16:23:11.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2021-02-25 05:38:28.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-26 02:07:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 10:29:22.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-02-19 11:13:07.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"225b2a791e00484b5a8f98044043a965b74f299f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-25 22:31:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a443b7afabf40acfc294d1d67b1eb551a1208ba","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-25 22:33:27.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"5f3b109261dbfbe07ecb395658139fb701c040ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 02:07:19.000000000","message":"Patch Set 1: Verified-1\n\n(26 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/dccccecddab54d95b54e3657b5522d84 : FAILURE in 50m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7d3851ee71be4df3a6f8a2c2c547b463 : FAILURE in 57m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e21cce5582e47698e737e60e8b42f77 : FAILURE in 18m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4765417cbab74e38b1d5cc6439b239a6 : FAILURE in 16m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b78cb330de744e3191505066c7b3c74e : FAILURE in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9492d0d122b4152977ebc7cf48ef78a : FAILURE in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37c4115b6c274b23a583889ef7ee9b4c : FAILURE in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2c0acc33aabe46e8aa83c5aadfbf7c99 : FAILURE in 4m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a8470222fce64cf0b89c524d3842d6d8 : FAILURE in 7m 18s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4c5b98b60d2f42b5b6e56643fc7ae6f9 : SUCCESS in 4m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7b58e1bbb38c4c64b8c741823d340507 : FAILURE in 19m 47s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4e2d337921b24475895896007efaf364 : POST_FAILURE in 55m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f4e951e58f7444b8a0841bb06bf3abdf : POST_FAILURE in 19m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/425e911bb01842d7a68aa42bab5b01d3 : POST_FAILURE in 21m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c6d2b495fc8407f8e20db459c4fa295 : FAILURE in 55m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/65c89a9551ea4c1d9cd22d5f02cac54b : FAILURE in 52m 48s\n\nWarning:\n  Comments left for invalid file glance/policies/image_tag.py","accounts_in_message":[],"_revision_number":1},{"id":"85d9d78cb4c4bc772f96d32ed85ce3c92cb9afa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-30 21:49:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f0e7cd7c9122d2686dcc559045b1776b128b26e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-30 21:54:10.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"262eefdf3a398a81a76ffe2573bc09dedf15c6f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 22:28:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d3fd8de780e44a3b94b3fe031ac89577 : FAILURE in 31m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/764cd5766db7409e917ae8a844a70a74 : FAILURE in 28m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab8ae5aee9f3443ab45871e36428149b : FAILURE in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05b4d0f6308f44ddaa44b24ac6963976 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/248e58550d4b4711ba5670671d803dca : FAILURE in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34b7ba1bb61e4edba5948a5759042a3a : FAILURE in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f137df9b55a94c8eb9a930614a3bebe4 : FAILURE in 7m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/61765e009acd4240a6c6a8590cc7fecf : FAILURE in 6m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/83df1166cb21423488311cc11311fe98 : FAILURE in 6m 35s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/57220d4938cc40e98d757fcb328940bc : SUCCESS in 4m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6bd9af71a513436bb2b67336576c8e66 : FAILURE in 28m 51s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e6236dbb195f49e1b50a0e242a5a7da7 : POST_FAILURE in 22m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/58c7ffb3bdfe4fc989e10922b12cafbe : POST_FAILURE in 28m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4bd117d6d36c4747b7c3ede86f6a3c75 : POST_FAILURE in 28m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/465c130ae5a242f38f47a16f5db55a9f : FAILURE in 16m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/647b6e755d2d48f6a3ad071891356d87 : FAILURE in 22m 11s","accounts_in_message":[],"_revision_number":3},{"id":"c58ce0c7c67e73443de3b498bbe49692a393243a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-19 10:29:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a95cbf2b3dbca77e0995ea6318f6916a6e9277fb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-19 11:13:07.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/764251/764251/4/check-not-vote/ext-glance-zuul/1d43b6e : SUCCESS in 41m 29s","accounts_in_message":[],"_revision_number":4},{"id":"11f8ff87a053d633912b24dd10e1c1f2acc46ab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 12:32:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/dd4251a41d014d3ab8ef4af7fc06c689 : SUCCESS in 1h 11m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a642275443e34056bf1c992cca5bdea0 : SUCCESS in 1h 28m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a620e4b2e1e944ee80c7e1e072d259d6 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8103256dc065459f834c81bfc2fd57ed : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eed6e2c2e92e49d49fdd2ec7d86e01cb : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e468878c5fe48ce9a14a1c6b9998d2e : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb5b9d8e501e44cf8b8221cf0aa92b45 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad748b02cdcf4af282126905387c7d9f : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5b82838f4fbd484da93c70a7730b0318 : SUCCESS in 8m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/74ea7edd387e4a139c0e6dede5c08679 : SUCCESS in 11m 04s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9763fbf0404948c8a6c6ab2d9491f450 : SUCCESS in 5m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/693df662c97f49768fb3649c892e3e76 : SUCCESS in 1h 19m 21s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3e79c741421c4c1681bd358557315325 : SUCCESS in 1h 09m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/07e91bbfd86a4d8b8239c65ce6a04d51 : SUCCESS in 1h 11m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1e2cf3d8495245b99e576f900119b8c3 : SUCCESS in 1h 15m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f998ac690cc416ba04eca6c349fb5f0 : SUCCESS in 58m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1796c1656cf64218894a3b68dba51ea8 : SUCCESS in 1h 19m 01s","accounts_in_message":[],"_revision_number":4},{"id":"2b22fa55861b0737903108e563d8923d4a40c91d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-19 15:08:23.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"f9cb1864485f81b1b147cc76d2512709972fac6a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-19 15:58:04.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/764251/764251/5/check-not-vote/ext-glance-zuul/393c564 : SUCCESS in 46m 22s","accounts_in_message":[],"_revision_number":5},{"id":"ed79bfea0103fd0681d64f4c3ce870d7344c8330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 18:17:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e693bdcb3c8841668ede1e60d34de01b : POST_FAILURE in 1h 09m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6b1b7d1ebac546d087210c3d8fe32a68 : POST_FAILURE in 1h 14m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b1cbac37e194ffeaf07972e4ea4e755 : POST_FAILURE in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8ba143cb61944dbbdf543b2e667787e : POST_FAILURE in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5e8e214ba954c6fb73609085049ebd9 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f8331f11b8f4b178382f2892e337f0e : POST_FAILURE in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d7d0c7bbe5f4c0997e941cde02326be : POST_FAILURE in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ac72cea77c4517860702b08faa63d8 : POST_FAILURE in 8m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4467e70d044841559b8c7338397afed0 : POST_FAILURE in 8m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c0b336ee25534fcab8608239d5e0b7cb : POST_FAILURE in 11m 17s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3e69d1974bff4319af900289f4d17f2f : POST_FAILURE in 6m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f4e0e671e4744eb89dc3033646fb0641 : POST_FAILURE in 1h 22m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/615e632987d94b9f81d1e5c57ae41fe8 : POST_FAILURE in 1h 14m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5262f02fff6f4f10b8aa481929b1ebb8 : POST_FAILURE in 1h 10m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d1a8925bc27843deb4060192851d54c3 : POST_FAILURE in 1h 15m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a54ed2703c24b9a9587d6f0c20e2fb9 : POST_FAILURE in 58m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5094f9991c2847bebf1e5d4ae895f14e : POST_FAILURE in 1h 05m 02s","accounts_in_message":[],"_revision_number":5},{"id":"a9723a5fc57a2d9ec504bc6e1928d04ce0162371","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-25 05:11:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f79384308af52b8aa98ebef2ab8c48fea7c6d122","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-25 05:28:23.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/764251/764251/6/check-not-vote/ext-glance-zuul/b80c05b : FAILURE in 15m 36s","accounts_in_message":[],"_revision_number":6},{"id":"d57dfd659d8fa0b8e6460bbe641b29e502217d61","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-25 05:38:28.000000000","message":"Patch Set 6: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6b9ffa7cecb7818e545256fc530684d23ae0a3ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 06:42:46.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b583ca33149a4608839104d4bdb2105e : SUCCESS in 1h 06m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3692a72b39744a1a968754de0252eeff : FAILURE in 1h 10m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f39c948d639240638d803bab808285b2 : FAILURE in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7bb4a6a6fd54963aa8807c291232122 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28c21c706c134a0389c88d2b86c0e6c0 : FAILURE in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91a86a25f66c410987b63ebfa99a8d0d : FAILURE in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac31c2d5edfe4abf9a456f3eb8f4e69f : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb1de2fd963f4f6f89989bcf8f38a1e6 : SUCCESS in 6m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7f204319730a4dcb99f27f61bec088b0 : FAILURE in 8m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9e9d40449b254542aa689295c97a654b : FAILURE in 8m 20s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/97810e784ea84bbf8ec0e606d78b8768 : SUCCESS in 4m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bf0e5d84bc644bae8dba385f109fc860 : FAILURE in 1h 15m 32s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f4d1b9ba5b394a00aaf116e75383c2f1 : FAILURE in 1h 04m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5b7216aa24394f66b6987735d213b68b : FAILURE in 1h 03m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c8a5e53f8b0042d3aeb137f90d83bead : FAILURE in 1h 29m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/667b243a3db2405a99c9ea3cdc90d26f : SUCCESS in 55m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bb6536c1e99444fb8fca5e86d47af379 : FAILURE in 1h 13m 33s","accounts_in_message":[],"_revision_number":6},{"id":"7716dd18a970a8df21c55f153587399c486f9045","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-25 18:04:15.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"5ce1d6aa7622f8cac0f750e1a8c65fa82aa752de","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-26 16:23:11.000000000","message":"Patch Set 6: Workflow-1\n\nMarking this as WIP until we figure out if this work is even feasible to accomplish for the entire metadef API in wallaby.","accounts_in_message":[],"_revision_number":6},{"id":"d786397f083617924a92e04abe3dd95f8a3b3f39","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-09-02 18:45:48.000000000","message":"Abandoned\n\nAddressed in different patch: https://review.opendev.org/c/openstack/glance/+/798700","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"eb547553e27cf8f42279501122e5b886c236de18","revisions":{"41bf977e9c4687203ff9dcff7040ee60f11140ad":{"kind":"REWORK","_number":1,"created":"2020-11-25 22:31:48.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/51/764251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/51/764251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/51/764251/1"}}},"commit":{"parents":[{"commit":"e60efa9068d6adde1a3ed7bdc08745fc13bad8cf","subject":"Implement secure RBAC for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e60efa9068d6adde1a3ed7bdc08745fc13bad8cf"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-25 22:27:14.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-25 22:28:18.000000000","tz":0},"subject":"Implement secure RBAC for metadef tags","message":"Implement secure RBAC for metadef tags\n\nThis commit adds two new policies for metadef tags. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC experience and\nimprove security.\n\nChange-Id: I33809f5fde7dd60b27156866744d2a06facd416d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41bf977e9c4687203ff9dcff7040ee60f11140ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41bf977e9c4687203ff9dcff7040ee60f11140ad"}]},"branch":"refs/heads/master"},"b2b19f2812809d242148dea0a2da0d22ba5c1cd5":{"kind":"REWORK","_number":2,"created":"2020-11-30 21:49:03.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/51/764251/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/51/764251/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/51/764251/2"}}},"commit":{"parents":[{"commit":"5f66b5234a704632a96e991182afb1bdcd6f5dc8","subject":"Implement secure RBAC for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5f66b5234a704632a96e991182afb1bdcd6f5dc8"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-25 22:27:14.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-30 21:44:50.000000000","tz":0},"subject":"Implement secure RBAC for metadef tags","message":"Implement secure RBAC for metadef tags\n\nThis commit adds two new policies for metadef tags. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC experience and\nimprove security.\n\nChange-Id: I33809f5fde7dd60b27156866744d2a06facd416d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b2b19f2812809d242148dea0a2da0d22ba5c1cd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b2b19f2812809d242148dea0a2da0d22ba5c1cd5"}]},"branch":"refs/heads/master"},"8ccd23c2b543ea7d677451fdb0f82bffaa28105c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-11-30 21:54:10.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/51/764251/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/51/764251/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/51/764251/3"}}},"commit":{"parents":[{"commit":"b2ddcbdc3aa975b4548fb096894a077281da4c7a","subject":"Implement secure RBAC for metadef properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b2ddcbdc3aa975b4548fb096894a077281da4c7a"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-25 22:27:14.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-30 21:49:50.000000000","tz":0},"subject":"Implement secure RBAC for metadef tags","message":"Implement secure RBAC for metadef tags\n\nThis commit adds two new policies for metadef tags. This is part of a broader\nseries of changes across OpenStack to provide a consistent RBAC experience and\nimprove security.\n\nChange-Id: I33809f5fde7dd60b27156866744d2a06facd416d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ccd23c2b543ea7d677451fdb0f82bffaa28105c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ccd23c2b543ea7d677451fdb0f82bffaa28105c"}]},"branch":"refs/heads/master"},"162d536b269b7488abdc8c861fc4f312b8d0bc15":{"kind":"REWORK","_number":4,"created":"2021-02-19 10:29:22.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/51/764251/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/51/764251/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/51/764251/4"}}},"commit":{"parents":[{"commit":"f3b79c917904733ad94098713f77467a63a29da3","subject":"Implement secure RBAC for Task APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3b79c917904733ad94098713f77467a63a29da3"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-19 10:26:48.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-19 10:26:48.000000000","tz":0},"subject":"Implement secure RBAC for metadef APIs","message":"Implement secure RBAC for metadef APIs\n\nThis commit updates the policies for metadef APIs to be default role\nand scope aware if ``enforce_new_policies`` option is set to True.\n\nNOTE:\nGlance is implementing Secure RBAC as EXPERIMENTAL in Wallaby, so to\nenable it operator needs to set ``enforce_new_policies`` to True along\nwith ``enforce_new_defaults`` of section ``oslo_policy`` to True in\nglance-api.conf file.\n\nImplements: blueprint secure-rbac\nChange-Id: I33809f5fde7dd60b27156866744d2a06facd416d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/162d536b269b7488abdc8c861fc4f312b8d0bc15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/162d536b269b7488abdc8c861fc4f312b8d0bc15"}]},"branch":"refs/heads/master"},"d764496192fd581d980285e40e82c4bfbb479b70":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-02-19 15:08:23.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/51/764251/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/51/764251/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/51/764251/5"}}},"commit":{"parents":[{"commit":"11093136787ab73c3faddabda4c0c5cff3b08bd9","subject":"Implement secure RBAC for Task APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11093136787ab73c3faddabda4c0c5cff3b08bd9"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-19 15:07:01.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-19 15:07:01.000000000","tz":0},"subject":"Implement secure RBAC for metadef APIs","message":"Implement secure RBAC for metadef APIs\n\nThis commit updates the policies for metadef APIs to be default role\nand scope aware if ``enforce_new_policies`` option is set to True.\n\nNOTE:\nGlance is implementing Secure RBAC as EXPERIMENTAL in Wallaby, so to\nenable it operator needs to set ``enforce_new_policies`` to True along\nwith ``enforce_new_defaults`` of section ``oslo_policy`` to True in\nglance-api.conf file.\n\nImplements: blueprint secure-rbac\nChange-Id: I33809f5fde7dd60b27156866744d2a06facd416d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d764496192fd581d980285e40e82c4bfbb479b70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d764496192fd581d980285e40e82c4bfbb479b70"}]},"branch":"refs/heads/master"},"eb547553e27cf8f42279501122e5b886c236de18":{"kind":"REWORK","_number":6,"created":"2021-02-25 05:11:54.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/51/764251/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/51/764251/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/764251/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/51/764251/6"}}},"commit":{"parents":[{"commit":"e14b0cde25ab721605f222d0dacfa119f02c5056","subject":"Implement project personas for image actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e14b0cde25ab721605f222d0dacfa119f02c5056"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-19 15:07:01.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-25 05:10:08.000000000","tz":0},"subject":"Implement secure RBAC for metadef APIs","message":"Implement secure RBAC for metadef APIs\n\nThis commit updates the policies for metadef APIs to be default role\nand scope aware if ``enforce_new_policies`` option is set to True.\n\nNOTE:\nGlance is implementing Secure RBAC as EXPERIMENTAL in Wallaby, so to\nenable it operator needs to set ``enforce_new_policies`` to True along\nwith ``enforce_new_defaults`` of section ``oslo_policy`` to True in\nglance-api.conf file.\n\nImplements: blueprint secure-rbac\nChange-Id: I33809f5fde7dd60b27156866744d2a06facd416d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb547553e27cf8f42279501122e5b886c236de18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb547553e27cf8f42279501122e5b886c236de18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
