)]}'
{"id":"openstack%2Fglance~766773","triplet_id":"openstack%2Fglance~master~Ifd55c44fef4e2187052d77084dc9c0fa9c9a0d16","project":"openstack/glance","branch":"master","topic":"lc_requirements_bump","hashtags":[],"change_id":"Ifd55c44fef4e2187052d77084dc9c0fa9c9a0d16","subject":"Bump lower_constraints and requirements","status":"MERGED","created":"2020-12-11 18:21:44.000000000","updated":"2020-12-15 00:41:29.000000000","submitted":"2020-12-15 00:38:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"766773-lc_requirements_bump","meta_rev_id":"4d88737f0d4e4ad6690c8fd055ac54e7f6986e44","_number":766773,"virtual_id_number":766773,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-15 00:38:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-12-14 22:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-12-14 22:55:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-12-14 22:55:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-11 20:08:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-12 01:39:54.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-14 05:10:37.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2020-12-14 20:02:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-12-14 20:59:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2020-12-14 22:47:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-12-14 22:55:34.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"53c3579b1840833100dcab263cdcc8198668cc7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-12-11 18:21:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"288a6b4f246f585e4f8f3e24ccc3bae11ac022d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 20:08:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/205410a7ae7148cf9d84a77f74189d1b : SUCCESS in 3m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/977e948cd2f54d8ab1ba60be2b14125f : TIMED_OUT in 40m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b5ba6c5cce946fe9021a8548925ab6f : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34757d38755849b482ef5b67ba694ff9 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c52fb267cfa442528a5b9e2f63cb406a : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2658dc201e124cff96952288e103e8bc : SUCCESS in 7m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9dff4895eca94372a1b02b6caf51a69b : SUCCESS in 8m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2216bd265ffa4d0f8d9c2c368417b3ed : SUCCESS in 7m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e6c12ae3c34d4c49b5d211b890b65d4e : SUCCESS in 1h 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"03bbf4b2b5a74eaf34514e4faa3d48966dd64a87","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-12 01:39:54.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/766773/766773/1/check-not-vote/ext-glance-zuul/c448734 : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"390d976123462c3a4d1b1d6704cf2e413ef6e640","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-12-14 05:10:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8875e2f9be13774cb794ae0e12cdcccc41c39bcd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-14 05:16:45.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/766773/766773/1/check-not-vote/ext-glance-zuul/2d73a7f : FAILURE in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ca56b575bc917763d3765659cbc9efd599ebcffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 06:18:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a61798868ce14217b02da50eaab1e1df : SUCCESS in 3m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fdf9e90d00264be78f862c6292685083 : TIMED_OUT in 40m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e67c04a216ba4abbb89fd778b41e90b9 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce2a801700f5487e9f6369240a6eed22 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1da709134d0648839da002dc9cfc1148 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eccbb96aa0544ce99be4077ddb322b3f : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f1c3b885f904435bbe16de9bb16dee52 : SUCCESS in 7m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/997d55fb14bf42b98d389e5e856a570a : SUCCESS in 7m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6a4123a9a3ce49d2998704032cc85f1a : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d25ee45f488079effc6230fa09c90f05e41c6cd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-12-14 14:16:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ffab73e0e370bbc1c56644d79ffd3709e738413e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-12-14 15:49:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ad9b99d1240594deb2ac9bf8482b74e252464f68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-12-14 19:14:11.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"1d7bb557b793b1740caf5020023d6db27ca00d0a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-14 20:02:16.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"79647610b30054dbe967c9a30904a2231869234e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-12-14 20:29:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4bab4bfc5170bf96b3f2941a31ccbaf9b148de5a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-14 20:59:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d76bcc155f397d88b243647371a104e9999eeeaa","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-12-14 21:30:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"01acc85d9fec8d05960690adbd7afd5f499606ac","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-12-14 21:33:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"034c35f027a65411c8fad3444899d7d748fac339","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-12-14 21:40:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"26c9ffa4c16c629f59f171fc1317f24447dc8d9c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-14 21:43:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"28909392b748b864d36816dcb15992b619d27d67","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-12-14 21:54:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9ca731073fe42939635c84f80989f35bff1ac679","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-14 22:47:03.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"50b48e08e8ef9b36d336cffc433bdf83c1c55b08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 22:49:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3c7f38e3b22d44ff9f365a6dd1d9662f : SUCCESS in 3m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aed8b28ebecd4fa0a82c79c782892201 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6da8e1f3af214418aca0c7991719bd33 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ecc8755bee649f6b42be65ddedb2114 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa802319698e4597bf199d9482bd54c8 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f672fb55348441b197fa2e627f96e81c : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/176b0a066c1945bbb09e0c5b20fca64d : SUCCESS in 8m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3db5853a318d474fa3a3ebbeee82b282 : SUCCESS in 8m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1a4fc0da27944a19a7e706b55b239690 : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":6},{"id":"1e803a4320081a68445e8ab77b0757452c522c38","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-12-14 22:55:34.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nThank you Erno!!!","accounts_in_message":[],"_revision_number":6},{"id":"5618fffb6d092b7503dccf1467662847b8851484","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 22:55:48.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7e3df6d967a7087975adf0e13f675143280e907b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 00:38:22.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/81e50c1d8eef44c5adf87f64e5134ab9 : SUCCESS in 6m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65669a6adc8444738dce462cf3f6efff : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/330f7deb7140445685fb75926684ac40 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6609fa0e8e45419f8fee8fe7de62371f : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89f02acb6e42425a99ef97d84a7119e5 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df73e2e0985c48bc99a26ccd8ddedf83 : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e7de21ea39bd4adf94ff00974e615eb9 : SUCCESS in 7m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dd8a4445110d489db049416f14e18961 : SUCCESS in 10m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c1d76a663f524432b8e1d2022d25e9b9 : SUCCESS in 1h 10m 14s","accounts_in_message":[],"_revision_number":6},{"id":"57ccd1209c4394af8ffefffd858c2c516b0fca59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 00:38:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"8e70372e9862bb3987e489807aaa42fcdc31ef66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 00:40:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/408daf9e29a346f6bcbc1b18355df612 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":6},{"id":"4d88737f0d4e4ad6690c8fd055ac54e7f6986e44","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-15 00:41:29.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/766773/766773/6/check-not-vote/ext-glance-zuul/3ceb100 : FAILURE in 4m 45s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"795dad763315884223b79698f3611e1ac1b50a17","revisions":{"0efa3b5f1f310d3e6ce6a1d9de3cc85cacf1d78e":{"kind":"REWORK","_number":1,"created":"2020-12-11 18:21:44.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/73/766773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/73/766773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/73/766773/1"}}},"commit":{"parents":[{"commit":"f102b74a285ad7de90c99ee7f5d6bef6e0d46161","subject":"Remove \u0027admin_role\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f102b74a285ad7de90c99ee7f5d6bef6e0d46161"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-12-11 18:20:33.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-12-11 18:20:33.000000000","tz":0},"subject":"Bump lower_constraints and requirements","message":"Bump lower_constraints and requirements\n\nThis is to unblock the stuck gate due to the lower_constraints\njob failing.\n\nChange-Id: Ifd55c44fef4e2187052d77084dc9c0fa9c9a0d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0efa3b5f1f310d3e6ce6a1d9de3cc85cacf1d78e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0efa3b5f1f310d3e6ce6a1d9de3cc85cacf1d78e"}]},"branch":"refs/heads/master"},"488f687bef1d4e72ad7c1aad429cb5025d01ab47":{"kind":"REWORK","_number":2,"created":"2020-12-14 14:16:08.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/73/766773/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/73/766773/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/73/766773/2"}}},"commit":{"parents":[{"commit":"f102b74a285ad7de90c99ee7f5d6bef6e0d46161","subject":"Remove \u0027admin_role\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f102b74a285ad7de90c99ee7f5d6bef6e0d46161"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-12-11 18:20:33.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-12-14 14:15:51.000000000","tz":0},"subject":"Bump lower_constraints and requirements","message":"Bump lower_constraints and requirements\n\nThis is to unblock the stuck gate due to the lower_constraints\njob failing.\n\nChange-Id: Ifd55c44fef4e2187052d77084dc9c0fa9c9a0d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/488f687bef1d4e72ad7c1aad429cb5025d01ab47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/488f687bef1d4e72ad7c1aad429cb5025d01ab47"}]},"branch":"refs/heads/master"},"6595f3b45993a7538450e8e3a09fb7c9fda68440":{"kind":"REWORK","_number":3,"created":"2020-12-14 15:49:32.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/73/766773/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/73/766773/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/73/766773/3"}}},"commit":{"parents":[{"commit":"f102b74a285ad7de90c99ee7f5d6bef6e0d46161","subject":"Remove \u0027admin_role\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f102b74a285ad7de90c99ee7f5d6bef6e0d46161"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-12-11 18:20:33.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-12-14 15:49:19.000000000","tz":0},"subject":"Bump lower_constraints and requirements","message":"Bump lower_constraints and requirements\n\nThis is to unblock the stuck gate due to the lower_constraints\njob failing.\n\nChange-Id: Ifd55c44fef4e2187052d77084dc9c0fa9c9a0d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6595f3b45993a7538450e8e3a09fb7c9fda68440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6595f3b45993a7538450e8e3a09fb7c9fda68440"}]},"branch":"refs/heads/master"},"245f793254980548166fc5a46e10ee20c51217ad":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-12-14 19:14:11.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/73/766773/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/73/766773/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/73/766773/4"}}},"commit":{"parents":[{"commit":"f102b74a285ad7de90c99ee7f5d6bef6e0d46161","subject":"Remove \u0027admin_role\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f102b74a285ad7de90c99ee7f5d6bef6e0d46161"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-12-11 18:20:33.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-12-14 19:13:52.000000000","tz":0},"subject":"Bump lower_constraints and requirements","message":"Bump lower_constraints and requirements\n\nThis is to unblock the stuck gate due to the lower_constraints\njob failing.\n\nDepends-on:\nhttps://review.opendev.org/c/openstack/requirements/+/767010\n\nChange-Id: Ifd55c44fef4e2187052d77084dc9c0fa9c9a0d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/245f793254980548166fc5a46e10ee20c51217ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/245f793254980548166fc5a46e10ee20c51217ad"}]},"branch":"refs/heads/master"},"d0ba4b8f3ebf42ca7b9e3595b8a0fcee2afaf8ab":{"kind":"REWORK","_number":5,"created":"2020-12-14 20:29:58.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/73/766773/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/73/766773/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/73/766773/5"}}},"commit":{"parents":[{"commit":"f102b74a285ad7de90c99ee7f5d6bef6e0d46161","subject":"Remove \u0027admin_role\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f102b74a285ad7de90c99ee7f5d6bef6e0d46161"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-12-11 18:20:33.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-12-14 20:23:14.000000000","tz":0},"subject":"Bump lower_constraints and requirements","message":"Bump lower_constraints and requirements\n\nThis is to unblock the stuck gate due to the lower_constraints\njob failing.\n\nAs per smcginnis on\nhttps://review.opendev.org/c/openstack/requirements/+/767010\nwe cannot drop the requirements on older than 3.6 python and\nthe requirements-check enforces the markers, our only option\nis to drop the job. Lower than PY36 is unsupported by Wallaby\nhttps://governance.openstack.org/tc/reference/runtimes/wallaby.html\nyet the requirements-check enforces it blocking us to move\nto supportable lower_constraints.\n\nChange-Id: Ifd55c44fef4e2187052d77084dc9c0fa9c9a0d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0ba4b8f3ebf42ca7b9e3595b8a0fcee2afaf8ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0ba4b8f3ebf42ca7b9e3595b8a0fcee2afaf8ab"}]},"branch":"refs/heads/master"},"795dad763315884223b79698f3611e1ac1b50a17":{"kind":"REWORK","_number":6,"created":"2020-12-14 21:40:47.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/73/766773/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/73/766773/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/766773/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/73/766773/6"}}},"commit":{"parents":[{"commit":"f102b74a285ad7de90c99ee7f5d6bef6e0d46161","subject":"Remove \u0027admin_role\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f102b74a285ad7de90c99ee7f5d6bef6e0d46161"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-12-11 18:20:33.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-12-14 21:39:58.000000000","tz":0},"subject":"Bump lower_constraints and requirements","message":"Bump lower_constraints and requirements\n\nThis is to unblock the stuck gate due to the lower_constraints\njob failing.\n\nChange-Id: Ifd55c44fef4e2187052d77084dc9c0fa9c9a0d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/795dad763315884223b79698f3611e1ac1b50a17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/795dad763315884223b79698f3611e1ac1b50a17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
