)]}'
{"id":"openstack%2Fglance~767076","triplet_id":"openstack%2Fglance~master~Icd82d3de1212992722c84f414d8ca4f8bdefdc22","project":"openstack/glance","branch":"master","topic":"bug/1886657","hashtags":[],"change_id":"Icd82d3de1212992722c84f414d8ca4f8bdefdc22","subject":"Global request ID erroneously interpreted as local ID","status":"ABANDONED","created":"2020-12-15 03:23:02.000000000","updated":"2022-04-15 07:11:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"842f73bb527a05eeab619607aae3e806b6b2b0a2","_number":767076,"virtual_id_number":767076,"owner":{"_account_id":32811,"name":"kwww1234","display_name":"K24","email":"kei24west@gmail.com","username":"K24"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-15 05:29:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 05:29:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-15 06:59:45.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"f501b1e9e9089f8f10c14d47152137fd74624096","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32811,"name":"kwww1234","display_name":"K24","email":"kei24west@gmail.com","username":"K24"},"date":"2020-12-15 03:23:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04d4b06fcd58c887f1dac32707851c14de8fa8ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 05:29:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/0bbbad9884854494a599fb0276b0d10f : SUCCESS in 56m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d69cd674d4e94afb890b71009b34c2b3 : SUCCESS in 1h 11m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cb356b4884042f8a8d7fb4c62048071 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c19a855bd441d7a28028fe20a93fa5 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0becab554671478f9040ada0b2e9130f : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8d8cb70e2734dc1a4eb8766045471c4 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53d18001b36a4f498aeffda6c501813a : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a405e9585fd74bd3a7e29bae630f7027 : SUCCESS in 7m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/12d15b27b05448d2ba1e5bfbeec3d701 : SUCCESS in 7m 11s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/840161540172480390de3d2d9c6fb791 : SUCCESS in 4m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/10c188f5ec5a47df99267599ba6a4fa4 : TIMED_OUT in 2h 03m 35s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3ff328713e264eb9be1f9fa923cdb72c : SUCCESS in 1h 22m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9a9d301828224ab0be732c33c41bea44 : SUCCESS in 1h 25m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/89f011353f234513b5858c7d0b2475c6 : FAILURE in 1h 21m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/066e2233870c4d949f3069dc9b5178e4 : SUCCESS in 1h 08m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c79850231bed42ad9e6fc5d05120fc58 : SUCCESS in 1h 15m 16s","accounts_in_message":[],"_revision_number":1},{"id":"259d76b43752a09b169d943cce342e1cb5b1bc78","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-15 06:59:45.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/767076/767076/1/check-not-vote/ext-glance-zuul/c9b27be : FAILURE in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"08a8897716c67dab08bda38d0eff2f93841ab904","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32811,"name":"kwww1234","display_name":"K24","email":"kei24west@gmail.com","username":"K24"},"date":"2021-06-07 14:54:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a012062fdcb4ab295a531dbf0cb27da8cec296f1","revisions":{"a012062fdcb4ab295a531dbf0cb27da8cec296f1":{"kind":"REWORK","_number":1,"created":"2020-12-15 03:23:02.000000000","uploader":{"_account_id":32811,"name":"kwww1234","display_name":"K24","email":"kei24west@gmail.com","username":"K24"},"ref":"refs/changes/76/767076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/767076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/767076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/767076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/767076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/767076/1"}}},"commit":{"parents":[{"commit":"795dad763315884223b79698f3611e1ac1b50a17","subject":"Bump lower_constraints and requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/795dad763315884223b79698f3611e1ac1b50a17"}]}],"author":{"name":"k24west","email":"kei24west@gmail.com","date":"2020-12-15 03:22:10.000000000","tz":540},"committer":{"name":"k24west","email":"kei24west@gmail.com","date":"2020-12-15 03:22:10.000000000","tz":540},"subject":"Global request ID erroneously interpreted as local ID","message":"Global request ID erroneously interpreted as local ID\n\nWhen operator specified UUID using \"X-Openstack-Request-ID\" header, OpenStack component logs it UUID as a global request id.\nBut Glance logs it UUID as local request id.\nBecause Glance assigned the received UUID to \"request_id\" variable.\nIt variable\u0027s value is logged as a local request id.\n\nFixpoint.\n・Changed the assignment of the received UUID from \"request_id\" to \"global_request_id\".\n\nThis bug may have existed since the Pike release.\n\nCloses-Bug: #1886657\nChange-Id: Icd82d3de1212992722c84f414d8ca4f8bdefdc22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a012062fdcb4ab295a531dbf0cb27da8cec296f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a012062fdcb4ab295a531dbf0cb27da8cec296f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
