)]}'
{"id":"openstack%2Fglance~768973","triplet_id":"openstack%2Fglance~master~I36e95dc0aa0698f75313718c0140c47a34f1f5db","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"I36e95dc0aa0698f75313718c0140c47a34f1f5db","subject":"remove py37","status":"ABANDONED","created":"2021-01-03 07:16:18.000000000","updated":"2021-05-04 21:29:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fde8b8d255308452f089f3f120f47ec9365efcc3","_number":768973,"virtual_id_number":768973,"owner":{"_account_id":31319,"name":"xuyuanhao","email":"xuyuanhao@inspur.com","username":"xuyuanhao"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-03 08:55:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":-2,"date":"2021-01-19 13:03:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":-1,"date":"2021-01-07 22:46:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-01-12 01:34:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-03 08:55:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-07 22:46:07.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-01-12 01:34:23.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-01-19 13:03:25.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"d14d1d6ed99943508683b716637a43566fabc9ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31319,"name":"xuyuanhao","email":"xuyuanhao@inspur.com","username":"xuyuanhao"},"date":"2021-01-03 07:16:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5bcd9972ea3e0ae99ce1cbbb84a68cb8230cbe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 08:55:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bf0f11b657fd461d9de4f9d8ca5fa317 : SUCCESS in 1m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/b659cc99f6d74b6d989c78923e68c352 : SUCCESS in 1h 06m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e3cea77c2625461eb753bc484051a571 : SUCCESS in 1h 05m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99477d22d95e460cac2d9d862217a12d : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/862febd58cf143889ece92eaf27f8a01 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cf6df45b54346769324681a5e6180ef : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23d6ba66bf4148f29cf5bd849f6282d3 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45415d0a0169420f948ac64da8921083 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4deff75068b945aa9870cd56aae0cc14 : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/251b84397332427a84cdb3214111a8ef : SUCCESS in 8m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/812deb9b28884a9b963377c172f9510a : SUCCESS in 7m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3bd02598c54a49ceabf9642ec5201645 : SUCCESS in 1h 02m 22s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b82a87fce0a1437b9d345af0e3afdc06 : SUCCESS in 1h 04m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/01d4eac647224aada20204fc722d5b86 : SUCCESS in 1h 18m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/36c8e3ed05e44e1d96d256edaf590cf8 : FAILURE in 1h 05m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a8f239627394e51a05b289a6bd09ec0 : SUCCESS in 1h 07m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9dcd49c51b144702bb1dbc99fb04a9e0 : FAILURE in 1h 37m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3347b48455fb601897005739b2969caa20e7acd7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-07 22:46:07.000000000","message":"Patch Set 1: Code-Review-1\n\nIt still support py37 as discussed in http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019648.html","accounts_in_message":[],"_revision_number":1},{"id":"4dd6ef3836216243711ad435813c69573d84d171","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-01-12 01:34:23.000000000","message":"Patch Set 1: Code-Review-1\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2021-January/019649.html","accounts_in_message":[],"_revision_number":1},{"id":"4831167501bc425352fd50768e3e3e62a22ac819","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-01-19 13:03:25.000000000","message":"Patch Set 1: Code-Review-2\n\nIndeed as per mailing list discussions linked in previous comments the 3.7 runtime is still supported.","accounts_in_message":[],"_revision_number":1},{"id":"fde8b8d255308452f089f3f120f47ec9365efcc3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-05-04 21:29:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"918162acd180684d26a91a5d58eec1a37f1e98c0","revisions":{"918162acd180684d26a91a5d58eec1a37f1e98c0":{"kind":"REWORK","_number":1,"created":"2021-01-03 07:16:18.000000000","uploader":{"_account_id":31319,"name":"xuyuanhao","email":"xuyuanhao@inspur.com","username":"xuyuanhao"},"ref":"refs/changes/73/768973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/73/768973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/768973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/768973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/768973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/73/768973/1"}}},"commit":{"parents":[{"commit":"73742be99944d923031aa5f90e06051126b17007","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/73742be99944d923031aa5f90e06051126b17007"}]}],"author":{"name":"xuyuanhao","email":"xuyuanhao@inspur.com","date":"2021-01-03 07:13:09.000000000","tz":480},"committer":{"name":"xuyuanhao","email":"xuyuanhao@inspur.com","date":"2021-01-03 07:15:54.000000000","tz":480},"subject":"remove py37","message":"remove py37\n\nChange-Id: I36e95dc0aa0698f75313718c0140c47a34f1f5db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/918162acd180684d26a91a5d58eec1a37f1e98c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/918162acd180684d26a91a5d58eec1a37f1e98c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
