)]}'
{"id":"openstack%2Fglance~770629","triplet_id":"openstack%2Fglance~master~Ica715fc1922f4b36dd0bb008ef6706b86115ec05","project":"openstack/glance","branch":"master","topic":"bp/distributed-image-import","hashtags":[],"change_id":"Ica715fc1922f4b36dd0bb008ef6706b86115ec05","subject":"Enable second glance worker for import testing","status":"MERGED","created":"2021-01-13 14:45:16.000000000","updated":"2021-03-10 02:30:59.000000000","submitted":"2021-03-10 02:27:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"770629-bp/distributed-image-import","meta_rev_id":"f82d3cf8e3483a1f1c4a5759c8c4badd4ed51e1a","_number":770629,"virtual_id_number":770629,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 02:27:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-03-04 20:25:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-03-09 21:06:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-05 13:32:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-03-09 21:06:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-03-10 02:30:59.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-13 14:45:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-15 08:45:12.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-01 17:22:44.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-03-03 17:12:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-03-05 13:28:04.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2021-03-05 13:32:19.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"518151cecc15e93604ae18f11f8f6a18e40c6218","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-13 14:45:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d948d8f1a5cf9ca32c9bedcdfbe0a31594ba50d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 14:45:40.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  while scanning a simple key\n  in \"openstack/glance/.zuul.yaml@master\", line 228, column 9\ncould not find expected \u0027:\u0027\n  in \"openstack/glance/.zuul.yaml@master\", line 229, column 7","accounts_in_message":[],"_revision_number":1},{"id":"ed5b4d7a87be3069344cf50796d96cb656404fab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-13 17:03:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cfc32fb4c8e6ad6730c61186733b6eca8889c6aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 18:22:44.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"bdc7cf182fb602c8ee76eb0de9b6e8025dc1bd2f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-13 18:24:14.000000000","message":"Patch Set 2:\n\nrecheck updated tempest test","accounts_in_message":[],"_revision_number":2},{"id":"47293f0d9937f54080fe0002deac32ca740ef132","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-14 00:28:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"62c90177d863a2653e4e8018c61bc593c6e04d97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 23:33:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1acaba61094d4366a9a564a517e2616f : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75726ff724c049a88f497f6971ac7566 : SUCCESS in 17m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad63673ee8554178979b49313a0576cd : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cab466f74f74415797a2566ee7cf1d64 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e649063c379432288a22f3cae3274fc : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72cb33aaac444483a74e34e242be08ac : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1d92f12bc3994ddda9221f7235656c6d : SUCCESS in 7m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/23ef778bd55f49b2906fe2f1f5fcaf79 : SUCCESS in 7m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bb009a77417d4b7bb2e9fa100c19bbb1 : SUCCESS in 1h 17m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/48e549740a884525b0c67c49f86dcf17 : POST_FAILURE in 1h 06m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/735a0e4e3ed14cebaba25e2071f04461 : SUCCESS in 1h 21m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f7d933c99ee471b68877d3cc9e3f21947f18bcbc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-15 08:45:12.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/770629/770629/1/check-not-vote/ext-glance-zuul/11d5980 : FAILURE in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"45beb8f2c7242caf737f552c37dd3e48ed0e0541","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-15 14:36:59.000000000","message":"Patch Set 3:\n\nrecheck infra failure","accounts_in_message":[],"_revision_number":3},{"id":"11143ec8d23ecd002eb00da174690b3687419919","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 23:24:33.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5ed29c2e14844668cac76cb1abc3643 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de82120cb0ac41909939606028b3dbd4 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a47296085494907b00f163b3b11d9f6 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1510a56ba75a48c1a9f91004d311ac8a : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/533959c66ccf431289b6b90cb0cbee9a : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be08f386679545488f0d2e3e1fda73e5 : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b4aef0a13f2c4b7193256417ad5c4203 : SUCCESS in 8m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64f8c5005d5b44a7a00dd04bc25f5199 : SUCCESS in 8m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b702496fa94d43e0a2a8ac0e7ef5f951 : SUCCESS in 1h 17m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/052d30589bbd426a86d9cf623f3c7c29 : POST_FAILURE in 31m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5bef7931f9a44d3cbf3c51632085e019 : FAILURE in 1h 15m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2d592a190785f5fc62971d1c96151e9d9363a8b7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-20 21:02:18.000000000","message":"Patch Set 3:\n\nrecheck infra fail","accounts_in_message":[],"_revision_number":3},{"id":"0517c81fa0536c1e896813b14743df128188eb39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 01:14:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/524234190cc84b028fe16d1f9e0cb810 : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b33742a4fd6646f1825c50a553340b2e : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/422d5ef5a2914357b197d089912af651 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a546e80e0bb47fc8499a0c3bc78f284 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c42d01163a74b21bf75d0c8deedc5d6 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13f98b0d61694566abbb50762a263926 : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4690c6d59b2d4ee6bf8ad385de990e17 : SUCCESS in 7m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b1458b22d3ac4a8aa2d1abccb1ae485a : SUCCESS in 7m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/88f5b628736f405bb32c86adfb5abb2f : SUCCESS in 1h 25m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b63b902cb8fd4d338d651d5764816b6c : POST_FAILURE in 33m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dd54f9619d3b475188078f360aeace12 : SUCCESS in 1h 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3915428f5daf5575546c211380532c8a460c8ca8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-25 20:08:28.000000000","message":"Patch Set 3:\n\nrecheck post failure","accounts_in_message":[],"_revision_number":3},{"id":"b42af4957ca5ad756a8d759a9d9052ec56ded764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 23:00:45.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ba98420384a48feb2431747300f6e68 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/701eb62b38e94377bf28010f24cb585c : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5099cdab0a6e45398e6b693c20b6b821 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1badea6198e74acbbbe055b7fc85e140 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf4fed9e90cb4a7cac5866d3237e93e4 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff636144282547b5a04f9896d73ce458 : SUCCESS in 8m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9633a1607a634dd7a68db21682f2b2ca : SUCCESS in 8m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dcf8e2a9e6cd43999d375e82353cf62b : SUCCESS in 7m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/36ed173663494d89851fa18e65cb3609 : SUCCESS in 1h 18m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/566806afdf914ac3b4180c86498a2050 : POST_FAILURE in 31m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/68823a847cd44a088fc2b08b28edb498 : SUCCESS in 1h 26m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"006aa54bba316b10fbda605fc7db1d8420c913d0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-28 20:58:14.000000000","message":"Patch Set 3:\n\nrecheck post fail","accounts_in_message":[],"_revision_number":3},{"id":"63bcbd3b3c6ee7ee178ab586972a31a432b1d7e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 01:09:24.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/054a962d0ead49bb84834c222cd72810 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ec4186a440d42a9b0000d73817af2a4 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/158825d8db83431896a9f11312458b17 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a40a487c90f4704815f2efee8270610 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3db11aa03c0447ba02fcc1bed2f4b88 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff2b306082b4a679f6f26b5daa04482 : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1fbcadab45384e2f944b68a0cdb3b06a : SUCCESS in 9m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ab4672fcfb204466912056c33b52fb0d : SUCCESS in 9m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/00f3ea9a0b56434fa53fc24a7c2f148b : SUCCESS in 1h 19m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dba2970f9a664179ab2438c315ebe681 : POST_FAILURE in 32m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5d42ea8193674e38ae6f5cd93e9d9122 : FAILURE in 1h 11m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"40e3c250714f30a267a1d954111878f8ccaded93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-01 15:35:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"541e4dabe05c38e7ccd3e312bc8d41b1ad747570","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-01 15:40:20.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/770629/770629/4/check-not-vote/ext-glance-zuul/c012563 : FAILURE in 4m 20s","accounts_in_message":[],"_revision_number":4},{"id":"a4ad659ef18634c6748d1966e833379b6a4d3937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 17:29:04.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"5f88aebb662a119b55c11a9d370d285fcf000c7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-04 16:10:51.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"0c815e2e980fc9e4ca81ee549698868545e1149f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 21:44:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a488f85ae8314382916e452230b3a422 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebec7ffcecf34c1286907295ebc73620 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/404bfb51552e4d2a98ff4876ad956bae : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d3f16ad6fa34a62906af00919e74551 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92c0404fe4294ac4b9574b68c5310ccb : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8174c7ab88d443dd95a9fe68bc99fd35 : SUCCESS in 5m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e935be70e7404453927b53fdd11dfb39 : SUCCESS in 8m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b5d542e8d4aa4e899da6aef6f22e5273 : SUCCESS in 6m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/83e68c60b4294a689f8ec6eb91cdb159 : SUCCESS in 1h 28m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b9f3dfc1b7b845e68ad4504b769f52f4 : SUCCESS in 1h 12m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7c71502b04fe47039b87f76fbe04fe56 : FAILURE in 1h 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f44c2976b059b29746047bb518ae8a70378be1a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-17 14:44:46.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"9cc2cf82c01e5b92a60f313ece0a372b19e06b28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 14:45:15.000000000","message":"Patch Set 6: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"3877fded6574ba8ab9aaec8dd21590c51b0b77e6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-17 15:55:34.000000000","message":"Patch Set 6:\n\nrecheck fixed merge conflict on tempest test","accounts_in_message":[],"_revision_number":6},{"id":"985eabf222f2d1d6d8b465b81114be31bf7948c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 18:52:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f992d50e0d545fbad34f4f61b01f1b1 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87c34e74814549ef8d025e4de754472e : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c217c1a36214a0da506c8152d16389b : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec09847e945c4d6bbbe73a023301188e : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3cab6c1833e41b88d129674255ca5f9 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa015ca829464bfe8d7a40d59f720f62 : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d8a8158204b64bb7b931460f33865e1a : SUCCESS in 8m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a9348008606b489f97d47d12f3cfe1ac : SUCCESS in 8m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0d607caa7e054d5b93051053718514b6 : SUCCESS in 1h 09m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a823d595964a40f29765835476b84f8f : SUCCESS in 1h 10m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8c458b5c86e64501801ff8ef2f5ddf36 : SUCCESS in 1h 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3419c2c6770ab403cd065abf63a051b1652ff52f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-01 17:22:44.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d13ec1b9b1bd0ba5e1dfda434ebe7692266ab8c4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-01 18:08:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"174600f46f12a7df869f79bebd5078e4f9ae2951","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-01 18:30:21.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"00803e571432cb264b90b9b4609ea02723dfd34b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-01 18:43:23.000000000","message":"Patch Set 7: Code-Review+2 Review-Priority+1\n\nNice to have this backed with tempest tests!","accounts_in_message":[],"_revision_number":7},{"id":"9d04c427644029703bd90a6308255887df900181","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-01 18:49:07.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/770629/770629/7/check-not-vote/ext-glance-zuul/5c90e92 : FAILURE in 17m 54s","accounts_in_message":[],"_revision_number":7},{"id":"868591ef4c0cd954cdcaf16eea70a24dd54c104f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-01 18:56:27.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"66983346be7999a3ac43096eb85fdc251cb6a38a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 18:56:47.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"b549c8476ae4051d16f95f36b96ef1bc8d4e8740","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-01 18:59:15.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"bb892fb376d7558ec765699de6672d78d19cb858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 18:59:46.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"e1babb62b3a1789ef0c4ec9e4624a9dbea2c41e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 20:46:26.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba8489a32f8b4377a8e779da4c725b1c : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/988bc4e97912466082f5eb9fc2997b66 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/773ec40bca574206a410b05ec058d483 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a5313253ffe4bee9a229c981250d3bf : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70b33fba9896444fa4f39615dbaa6e8d : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc0726989de345aea4102db67c128d5f : SUCCESS in 6m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf84e9ca0bf4451b9984bfffd4268079 : SUCCESS in 7m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4e75d839c04d4fcfba72e9869a35350a : SUCCESS in 8m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/afdecae1c56e4e17bfe232af6bae0131 : SUCCESS in 1h 12m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7546e322302248a3ae3473ac15ab2cf4 : SUCCESS in 1h 09m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1c7b269d75fa45bf9fae8a216cba8f9c : SUCCESS in 1h 07m 41s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6437bc374c2914a5ddb80360924a3e38b78e2594","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-02 19:53:18.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"5e1ef9999bab470e97c65da47eb75077e0082c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 21:21:05.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d7240174ea244cbb9e11de59192ca4f : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd69df3bf7c4c3fafc606f3c0658edc : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6122190f5cbe4fbeb5cd14f43a137317 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/436331d2f4754b369a0e7ceaf8e86444 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cc99b9844af4f738a8d3ff9ce4fab9a : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c8dd6dd32f143148e5d6101a8f0ec91 : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/50e546e5ddad4abfa65d50cccb7e57a5 : SUCCESS in 8m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64469e50dd154ab584694b54c8c13b8e : SUCCESS in 14m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4e89d6cf8690468e9558e5a904a014cf : SUCCESS in 1h 15m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9ba1e54263d44e3eb56f83803a41d417 : SUCCESS in 1h 03m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/202aa9143de44b3c853d8206b0ebb242 : SUCCESS in 1h 20m 38s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f63ef207500604b2ed6c590b267a214aa1625465","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-03 14:37:56.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"032a60cb45c582205eebdc847122e7ec231d17c4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-03-03 17:12:52.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\nAppears to be happy with this.","accounts_in_message":[],"_revision_number":11},{"id":"1ba885e57f3949e7004c2b8087098f1571322494","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 19:55:46.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48ecbfafc3914d0ba99d092c21438f3e : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88e8ef76adcd4eacb1ad4d63e27edf2a : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/338ba0b6dea34463b98160ff7c48d1fc : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08779d4b9a364711aa501abf1af528e4 : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a44e993fb4a4aa9a29c5f35e8b13034 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db7b9dd0a81e4d488f10511a14307dce : SUCCESS in 8m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7798c59b20924582b86bec49852a57dd : SUCCESS in 8m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/20f64f1ebd05405abb29050bf377f208 : SUCCESS in 9m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a17cfbeb1606450ebe8044ed610f215c : POST_FAILURE in 41m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fa214c25d8f5444ea58e20706f818037 : SUCCESS in 1h 16m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/edd1a4e72c3e42229642314bdacbce96 : SUCCESS in 1h 24m 24s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"141bb78e623bfac620884c0addbf7561625343a9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-03 20:52:42.000000000","message":"Patch Set 11:\n\nrecheck g-api-r restart supposedly failed, but doesn\u0027t look like it in the logs","accounts_in_message":[],"_revision_number":11},{"id":"67488404133b615c737ff89c0ceacf1d54deb838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 22:57:55.000000000","message":"Patch Set 11:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"eb5159372e83b53da35c041dda33142b930a16a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-03 23:18:32.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"11e38afaebda516a56c6e2098891689f0017b149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 23:29:01.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"1d084ef5bce9c4cd9c261aa9716c1a2787af199a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-03 23:29:53.000000000","message":"Patch Set 12:\n\nrecheck devstack patch changed","accounts_in_message":[],"_revision_number":12},{"id":"12fe0a74f10ea6e7a1a26fb0f3f93cb31b312990","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-03 23:33:53.000000000","message":"Patch Set 12:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/770629/770629/12/check-not-vote/ext-glance-zuul/2080813 : FAILURE in 14m 20s","accounts_in_message":[],"_revision_number":12},{"id":"5177ddef4ab04c8a9779db0aa701277a04cc2778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:58:46.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/867037d70b0c40e2958dd3049d2ebbcf : FAILURE in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8bea6381489488ab69188432387611c : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a82551e84d5540aca4493700a75bcb69 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/066607f974484202a321fec010a68edc : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e4af15dc06b403c9bf381aa4266281f : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6922ef2fefdd4385bccf2c8f25514133 : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d873cb4f73d64023834427a00842d557 : SUCCESS in 9m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a16c4a0db2804e20855ff8276be16e6f : SUCCESS in 9m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4ff68c515fe04ee6aef629ebee0e1024 : SUCCESS in 1h 22m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ac091aa3ded14878a2b099ca89b337f6 : SUCCESS in 1h 15m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bdb70acc32ac4ea88ac55b83eb5122ba : SUCCESS in 1h 17m 10s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6da0ff7e29d982db35ccc8b23e1e16c24b427559","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-04 01:14:01.000000000","message":"Patch Set 12:\n\nrecheck looks like unit test abort or something","accounts_in_message":[],"_revision_number":12},{"id":"54f4f1c329f4b72df07f01c233873e9142ee3bb0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-04 01:32:51.000000000","message":"Patch Set 12:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/770629/770629/12/check-not-vote/ext-glance-zuul/c36b384 : FAILURE in 18m 05s","accounts_in_message":[],"_revision_number":12},{"id":"12aeb2cb89840b6dd2648aa2806d71e3dd07b903","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 02:52:10.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afa3d59881b045d08fae85f108baa5ca : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cd7b6f0e9054079b9f042ffd6344f4f : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f971c572f3fa4c8e9d0a6d0bef75718e : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/950e46d821314071b7aa34a7f584f565 : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12d89bc803454d1baaf54d2d1a551da3 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a50ce86f0554240a894c041b2c3522d : SUCCESS in 7m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9d615b0d6f174ad0896438252f64b1da : SUCCESS in 9m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/50feb388049947be961283d1bafc9bdc : SUCCESS in 10m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0aa1ea7233b648faa51a18944a48f7c8 : SUCCESS in 1h 31m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3342231bdd2a4e949f5d186b395cc864 : SUCCESS in 1h 35m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/43dfcbaa934f483bac0ce10bd6545ca7 : FAILURE in 1h 34m 08s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5a2e407250499a7435c9189712e291beaab31131","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-04 03:13:28.000000000","message":"Patch Set 12:\n\ntest_remote_delete[id-44d60544-1524-42f7-8899-315301105dd8]\n\tpass\ntest_remote_import[id-e04761a1-22af-42c2-b8bc-a34a3f12b585]\n\tpass","accounts_in_message":[],"_revision_number":12},{"id":"e31f7eb7b446bd511e9a7105a1fe88c1a7763060","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-04 03:13:42.000000000","message":"Patch Set 12:\n\nrecheck run this again just to make sure that g-api-r restart thing was a fluke","accounts_in_message":[],"_revision_number":12},{"id":"a88f7d9b1f2f350c35d6d106adfc97272c48e78b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-04 03:27:15.000000000","message":"Patch Set 12:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/770629/770629/12/check-not-vote/ext-glance-zuul/6512c6c : FAILURE in 12m 33s","accounts_in_message":[],"_revision_number":12},{"id":"c9bd806a1b28ffeedef949e842e423768bfedf12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 04:56:32.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50a50a4c33f347ecb04d695dc2a5d74b : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cfb6fa3292045fdb6225426f49f7892 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58a9eb9c43bb4c02b1c2ab9f11f062d1 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ab79c5303d447f3b42bfaef7b8f602a : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/656f2cd75e5f49d79328db543f94731d : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88b5580d9a494869bb6641578c4de0c6 : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/703a091d735f43ccb3b969c14abf5296 : SUCCESS in 9m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/411a33de452e4a5391b3d58f6e022448 : SUCCESS in 9m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0e52681ba34c42208dd086cc37f02157 : SUCCESS in 1h 03m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ebf0cb95d0674d66bb23e579cd5676f0 : SUCCESS in 1h 40m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f7d60ba0858b4a00a4af69ade787ad8d : SUCCESS in 1h 13m 27s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"562e2359b9855295c8e542477f662f8cfda36115","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-04 05:41:24.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)\n\nJust for FYI, waiting for dependency patches to merge before I put +2","accounts_in_message":[],"_revision_number":12},{"id":"bc76008201d94d4d44f71afa781c0c323f8d9d1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-04 16:48:36.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"ec43289ca84d97d637b18157a46d4a00358b7584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 20:03:59.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e528339b226a47c8bcaa35a609bd067d : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d56c000c6a354c68b03bc77f571ce392 : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c53415e9663c464c9f74e5801e3c76ea : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60656b4e68ed423b85095de8a24e1bce : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd760bdbe43a4883ad6efd4e1acb80c5 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e872bd8a846422faa12e782fa62ed2c : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f21b1506e55241bbbd72cd499579189b : SUCCESS in 10m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/69de7765ebe64b6e8f607944d103cf87 : SUCCESS in 16m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/97e01906f0b04e2ba53019d36181f329 : POST_FAILURE in 1h 05m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8eed8cf1f9fb47538eea2bceea96670a : SUCCESS in 1h 10m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a1e518dcc81c49838d0d66f38655fada : POST_FAILURE in 1h 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"5ff08f48e5601e7f36bc855bc80abeb347de0c25","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-04 20:12:44.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"5e94d59555feed0d94f351f28d914519c6ea23d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-04 20:13:28.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"98484984240cc7cdeb1ab2236248113c980ff26d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-04 20:25:00.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"2dd8f1fb5de99b1a47733b8e8a614dc60f23b411","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-04 20:25:32.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"db7cf8db9ef63e257c34480200cbcd8d6d747711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 21:38:52.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cea777e1d23841ea923ffd65f96c78cf : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81c3dca336a44275ba554586b0aa67e0 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ac99551ce00428a8e111935d0bb87a6 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee60ca29c55a41bfa3591ef365a8297e : SUCCESS in 8m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/505afc5a56e14eaea0f7ff207e4bccce : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fd498d5370d466b9f86fd5ff331e180 : SUCCESS in 8m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/055d7f22503d4997a5fbcffdce6ff693 : SUCCESS in 9m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d05214a207f74756972ef8e457b0b776 : SUCCESS in 8m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/837c5c3437c14e518b0a5f6b5e84ebcb : SUCCESS in 1h 21m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e8b154c959be4b4cb2dc521156a9de3d : SUCCESS in 1h 18m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4f97ef4fd67a4d9a90309e9ac874d573 : SUCCESS in 1h 09m 13s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"becb1235942b7e51356f0e4ad19dc3a43d3f59dd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-05 06:32:15.000000000","message":"Patch Set 14:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/770629/770629/14/check-not-vote/ext-glance-zuul/abf8294 : FAILURE in 35m 24s","accounts_in_message":[],"_revision_number":14},{"id":"60a10a6af51aa8f54ea77b533965f7235b88398d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-05 13:28:04.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"454c3bd7bc029d8e740de311ea100838255d3ee5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-05 13:32:19.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"c953ec65ec264dbe414f84c431761a2864cef324","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-09 02:27:13.000000000","message":"Patch Set 14:\n\nrecheck devstack changed","accounts_in_message":[],"_revision_number":14},{"id":"af4206ca0010a7de9a0e5952502ecfdbf94273b6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-09 02:57:47.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/770629/770629/14/check-not-vote/ext-glance-zuul/2cc98ce : SUCCESS in 29m 37s","accounts_in_message":[],"_revision_number":14},{"id":"fa3215314251b0ad7212ef966a21b18ab1edc5b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 03:49:12.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0fdb6eaba4c4d609295004fa73c58fa : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfde018845264400b46b45b3ebd8c6d7 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98e178ded41c477d9cd4757166a14df8 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bd75f312a5f4148a2442dd2f6274fe3 : SUCCESS in 8m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ebd0d8391f548bd95f70e4b0ab8beb1 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2707929235ab43eea2aa88259011c0a7 : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6a0b4b20628045a5a70d55e5d9a08686 : SUCCESS in 8m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/510a2c13614942ffa1294d1c3c4d2004 : SUCCESS in 7m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1892ab89de7543e0b9be7fd2f0b149ce : SUCCESS in 1h 14m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c9866f7382ef470bad9900c5336148fc : FAILURE in 1h 19m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/90ee084be6ca44d9a3e2f4fd23e177f4 : SUCCESS in 1h 19m 38s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f8106be27d434471baea1d85a821b8bd : FAILURE in 34m 44s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4e3de712e5ce477b87a0a64798d03fe6 : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":14},{"id":"849840366d8b90feaaf58bc84d0c01b298a7f641","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-09 03:50:38.000000000","message":"Patch Set 14:\n\nStill works with devstack changes:\n\n test_remote_delete[id-44d60544-1524-42f7-8899-315301105dd8]\n\tpass\n test_remote_import[id-e04761a1-22af-42c2-b8bc-a34a3f12b585]\n\tpass","accounts_in_message":[],"_revision_number":14},{"id":"66000fd487c8cb4ef8055e4122eacbb8ca05e130","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-09 03:50:43.000000000","message":"Patch Set 14:\n\nrecheck nova fail","accounts_in_message":[],"_revision_number":14},{"id":"86c0ab751aaa8cdddcf1b9a62c9cf81a7bbe7074","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-09 04:28:44.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/770629/770629/14/check-not-vote/ext-glance-zuul/b459368 : SUCCESS in 37m 08s","accounts_in_message":[],"_revision_number":14},{"id":"ff01d24aea57af82341571c3e036f33b031db0ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 05:04:39.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2898ae6e8fc6425a857084ef1d408228 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e29ba35454e24303a8db189ceb41b24c : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afb42a6d391c46b886eb5a7c70fcf83b : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/489196be13ba4d329ceb1164272ec913 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cddffe8a09314dc583ca827bd3901495 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/253bdbffacad4d4597df24dec9b257e3 : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ca43338d4f4445b08867bdf89d5958b9 : SUCCESS in 8m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/19e7348b67314ecfb3ffa553e958fea2 : SUCCESS in 8m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/41f53ebba7e14dd0b048658f01116888 : FAILURE in 59m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/93fa743fcad243488ad64fc3fcc22f2f : SUCCESS in 1h 12m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fbb46536d3834719994cdee7b3dcb052 : SUCCESS in 1h 09m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2a92093dfa9a418094ad713d5776999d : FAILURE in 28m 47s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/95cf2128becd4f109d81e87a28a1f3dc : SUCCESS in 30m 15s","accounts_in_message":[],"_revision_number":14},{"id":"f67e5f5fef33cc2ba51a4da0b56afddfd2ab2f97","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-09 05:12:33.000000000","message":"Patch Set 14:\n\nrecheck cinder failure","accounts_in_message":[],"_revision_number":14},{"id":"a14a8eb66fa9b475ad02e278ba6f7212f38642cf","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-09 05:52:53.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/770629/770629/14/check-not-vote/ext-glance-zuul/2772303 : SUCCESS in 39m 32s","accounts_in_message":[],"_revision_number":14},{"id":"2e6cf9b18cdaedf6b264171feed14ddcf54c8537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 06:43:20.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/448935dcd48543b6bcf88b0ea10da648 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06a6f1b6be654fe2a05a3f97f1bb09b5 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf16ce73153749e5a76212476fa65de5 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa51672138dc4d4c9f71ef8aaf81d9fc : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a0735bdd9894220a1deae58bd58b822 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77fc967e272c48db9806e5941015c99a : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/592a9e169d174ccd90aa98acd06f31da : SUCCESS in 7m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c8762989e76f4292966502fca1ad01d9 : SUCCESS in 7m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/88894895a7284edca893d2778ede2949 : SUCCESS in 1h 15m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/83e60dc87f0d424897a154decb66ca67 : FAILURE in 1h 08m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b35dc9aea0de4045b941951f5c55dce9 : SUCCESS in 1h 24m 55s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bebaf61779654b3598d7506a9e9fa0b2 : FAILURE in 33m 58s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/327a76ec147348849e957d59577234bf : SUCCESS in 39m 28s","accounts_in_message":[],"_revision_number":14},{"id":"785256ad818c86ba8cbdea110e98472135aa533d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-09 07:10:02.000000000","message":"Patch Set 14:\n\nrecheck cinder failure","accounts_in_message":[],"_revision_number":14},{"id":"2f77807a7f928727cd8e2097e1b8c75d7a864003","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-09 07:45:45.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/770629/770629/14/check-not-vote/ext-glance-zuul/6cc1050 : SUCCESS in 35m 04s","accounts_in_message":[],"_revision_number":14},{"id":"13d91da1da32b50a0f55ad0187559724b9880d85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 08:29:52.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d40ec4a498dc4760821572a7dbdc0113 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d74ddc4c73734d088421fd86d4e361c3 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e00b5082df348cbad71ca5f2961bf5d : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1083b0af3a7f4687ac2c37f7bf3ab4ac : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/108611aae5d24670ae8d208e867395bd : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2a2d4458057411a872e2ec3fbd3eb81 : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1609f497ba614fe7bbcdb72e136f4b00 : SUCCESS in 8m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b257a98c4e0149c7a9e93c44049872dd : SUCCESS in 8m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0c0969e1f70148c19e556708f0f68783 : SUCCESS in 1h 16m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/004f0709df2040b49b1be9693f62ab7e : SUCCESS in 1h 07m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bef93ce8f60a4f30ba8867c45b8fda14 : SUCCESS in 1h 18m 23s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/370d75cc2a5f4cb48c98892280be300b : FAILURE in 32m 55s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ee7a57e0928342de924c2bbc33deee3a : SUCCESS in 31m 58s","accounts_in_message":[],"_revision_number":14},{"id":"b7790d0c1898e6df09a7dacfbd8a79d840037c04","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-09 21:06:07.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\nRe-approving this now that dependencies are merged","accounts_in_message":[],"_revision_number":14},{"id":"d2c4183380597a0151a47f231cd3c0496327bb58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 21:07:07.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"9161dbf49e59ed9b42fa14595a579ff07ce8ddec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 23:34:56.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad4a4a9b0f4c4b349a1795b12c599f8d : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db6f6ff4dfdf498197843aa5adf0ae34 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14a4fa768b454a1183e634c0dbc72be7 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7c6564862614f02913b1b5dfad0905d : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b170ae8ba584267ad9bf996b21b20c7 : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d99fcf04b8464112aede5b63d08a9d17 : SUCCESS in 9m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/961a6cab596e4341ba13fadc60dec6e9 : FAILURE in 11m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ba0c83b1637e4cc6b45738836e1a4f34 : SUCCESS in 1h 07m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0c366f2ba28249999db12a73e5a066c5 : SUCCESS in 1h 06m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ba1be9d148744a8a6348386c6c5e0df : SUCCESS in 33m 43s","accounts_in_message":[],"_revision_number":14},{"id":"4856221bb2fccf4fcce4a43b876e05a95195d73d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-09 23:38:44.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"6e5d53f9fb9be60125d4b136b8b86fec98316837","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-10 00:16:14.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/770629/770629/14/check-not-vote/ext-glance-zuul/62322aa : SUCCESS in 36m 45s","accounts_in_message":[],"_revision_number":14},{"id":"e78d371f82e45933883caefc36fb1c679f1c648f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 00:59:49.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf833210229544f5bd3667ba4364652d : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/606098551ea94954ba6dd50e31757a30 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/897903c6b27a40ddb614b314e83d5c44 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79dc936e45e748c0ac0bca56587eb823 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fe75caa426d4cb0805e9b97ccf7c7f7 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f0923fb1644b019870d25fd3635b07 : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/384124f09ef044bb8dc69335f0703811 : SUCCESS in 10m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4670458fbaf24f078a5c9ae649eba7f2 : SUCCESS in 9m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c688c7a6c5b84f59bb076bc59a981d7e : SUCCESS in 1h 15m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c4caeb9ecc1b4125a4b912a5d383f49f : SUCCESS in 1h 16m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9f7d740d230b48e0bebeba10c2d76429 : SUCCESS in 1h 19m 13s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bc6ab09702724dc498ab98723f483b1e : SUCCESS in 33m 18s","accounts_in_message":[],"_revision_number":14},{"id":"6f4259132fe7321bc6563ede1ac40520b0be6020","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 01:00:29.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"19b185b8473a22993f08bdae862d048a1d4affa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 02:27:06.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0685c11a9fb4450c873a7291d1476dbf : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/050dde79354d43b0862b8abd93d6c17e : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b2f0dfe4fbe4125a874377cbf6cde8f : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/763e3090dc124d08ba44f06b5474e98e : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24da20a2bb0c4a30b11292b4c7634371 : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2dfafd64469c40c1b5ec59b9628d3c85 : SUCCESS in 8m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/089d8fbffcde42f297ddd01cbf2aa660 : SUCCESS in 8m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/00c26606c0e14318a550e787746855b8 : SUCCESS in 1h 11m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9b440666b63c427db07d2c8ca222d579 : SUCCESS in 1h 21m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b2203aa200154a75ae61b0be5ecff197 : SUCCESS in 28m 24s","accounts_in_message":[],"_revision_number":14},{"id":"0bab38bd96a074a037002fdc7b3cd698675620a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 02:27:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"7865567b06597613f72df4117789da44e2be773c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 02:30:59.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0fb5afc901e491da97144baad357923 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"662607f496bf0267b265a0820a91783c15270ea4","revisions":{"7d8476dc6e734b971178e42d6acd29d9b519d7b7":{"kind":"REWORK","_number":1,"created":"2021-01-13 14:45:16.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/29/770629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/770629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/770629/1"}}},"commit":{"parents":[{"commit":"eeb0b1c86a1349a7ee73c07f268a88672796fe7b","subject":"WIP: Distributed image import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eeb0b1c86a1349a7ee73c07f268a88672796fe7b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 15:41:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 15:41:35.000000000","tz":-480},"subject":"Enable second glance worker for import testing","message":"Enable second glance worker for import testing\n\nThis enables the g-api-r service in devstack, which allows tempest\nto run the remote import test, causing it to stage and import an\nimage across two different workers.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/770487\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/770520\nChange-Id: Ica715fc1922f4b36dd0bb008ef6706b86115ec05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7d8476dc6e734b971178e42d6acd29d9b519d7b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7d8476dc6e734b971178e42d6acd29d9b519d7b7"}]},"branch":"refs/heads/master"},"3660bbb71f25b268f0528f32ef5597e16b4337ae":{"kind":"REWORK","_number":2,"created":"2021-01-13 17:03:28.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/29/770629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/770629/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/770629/2"}}},"commit":{"parents":[{"commit":"f5ffbbe90610d8a30f6e471191e932065662ac75","subject":"WIP: Distributed image import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f5ffbbe90610d8a30f6e471191e932065662ac75"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 15:41:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 17:56:29.000000000","tz":-480},"subject":"Enable second glance worker for import testing","message":"Enable second glance worker for import testing\n\nThis enables the g-api-r service in devstack, which allows tempest\nto run the remote import test, causing it to stage and import an\nimage across two different workers.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/770487\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/770520\nChange-Id: Ica715fc1922f4b36dd0bb008ef6706b86115ec05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3660bbb71f25b268f0528f32ef5597e16b4337ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3660bbb71f25b268f0528f32ef5597e16b4337ae"}]},"branch":"refs/heads/master"},"2d6f631fb79ec39f0f71941bae38a2186feb78ce":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-14 00:28:41.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/29/770629/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/770629/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/770629/3"}}},"commit":{"parents":[{"commit":"14b15031a9a1aa85ed14c725ebef6d4d52419fcb","subject":"WIP: Distributed image import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14b15031a9a1aa85ed14c725ebef6d4d52419fcb"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 15:41:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-14 00:45:38.000000000","tz":-480},"subject":"Enable second glance worker for import testing","message":"Enable second glance worker for import testing\n\nThis enables the g-api-r service in devstack, which allows tempest\nto run the remote import test, causing it to stage and import an\nimage across two different workers.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/770487\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/770520\nChange-Id: Ica715fc1922f4b36dd0bb008ef6706b86115ec05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d6f631fb79ec39f0f71941bae38a2186feb78ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d6f631fb79ec39f0f71941bae38a2186feb78ce"}]},"branch":"refs/heads/master"},"bb01756215e16d74fb09d8fc9342eff387a2e122":{"kind":"REWORK","_number":4,"created":"2021-02-01 15:35:01.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/29/770629/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/770629/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/770629/4"}}},"commit":{"parents":[{"commit":"14b15031a9a1aa85ed14c725ebef6d4d52419fcb","subject":"WIP: Distributed image import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14b15031a9a1aa85ed14c725ebef6d4d52419fcb"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 15:41:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-01 16:32:41.000000000","tz":-480},"subject":"Enable second glance worker for import testing","message":"Enable second glance worker for import testing\n\nThis enables the g-api-r service in devstack, which allows tempest\nto run the remote import test, causing it to stage and import an\nimage across two different workers. Note we disable it for the\nstandalone mode, since devstack does not support starting another\nstandalone glance.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/770487\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/770520\nChange-Id: Ica715fc1922f4b36dd0bb008ef6706b86115ec05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb01756215e16d74fb09d8fc9342eff387a2e122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb01756215e16d74fb09d8fc9342eff387a2e122"}]},"branch":"refs/heads/master"},"415309e099c727268afe3f3698a2af82696c41a5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-02-04 16:10:51.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/29/770629/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/770629/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/770629/5"}}},"commit":{"parents":[{"commit":"059a161ff080ab1db47c991dcb5ab18cb71f4f1d","subject":"Distributed image import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/059a161ff080ab1db47c991dcb5ab18cb71f4f1d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 15:41:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-04 17:05:49.000000000","tz":-480},"subject":"Enable second glance worker for import testing","message":"Enable second glance worker for import testing\n\nThis enables the g-api-r service in devstack, which allows tempest\nto run the remote import test, causing it to stage and import an\nimage across two different workers. Note we disable it for the\nstandalone mode, since devstack does not support starting another\nstandalone glance.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/770487\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/770520\nChange-Id: Ica715fc1922f4b36dd0bb008ef6706b86115ec05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/415309e099c727268afe3f3698a2af82696c41a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/415309e099c727268afe3f3698a2af82696c41a5"}]},"branch":"refs/heads/master"},"508cc6d37b0101e3308c90056814bd13549b3699":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-02-17 14:44:46.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/29/770629/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/770629/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/770629/6"}}},"commit":{"parents":[{"commit":"bc0525f365c6be078506179b4bafa976ce9c1f0a","subject":"Distributed image import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc0525f365c6be078506179b4bafa976ce9c1f0a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 15:41:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-17 15:43:22.000000000","tz":-480},"subject":"Enable second glance worker for import testing","message":"Enable second glance worker for import testing\n\nThis enables the g-api-r service in devstack, which allows tempest\nto run the remote import test, causing it to stage and import an\nimage across two different workers. Note we disable it for the\nstandalone mode, since devstack does not support starting another\nstandalone glance.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/770487\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/770520\nChange-Id: Ica715fc1922f4b36dd0bb008ef6706b86115ec05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/508cc6d37b0101e3308c90056814bd13549b3699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/508cc6d37b0101e3308c90056814bd13549b3699"}]},"branch":"refs/heads/master"},"447602a703974a371f0f2d8e50dd12af8a12c9f6":{"kind":"REWORK","_number":7,"created":"2021-03-01 18:30:21.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/29/770629/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/770629/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/770629/7"}}},"commit":{"parents":[{"commit":"bc0525f365c6be078506179b4bafa976ce9c1f0a","subject":"Distributed image import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc0525f365c6be078506179b4bafa976ce9c1f0a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 15:41:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-01 18:30:17.000000000","tz":-480},"subject":"Enable second glance worker for import testing","message":"Enable second glance worker for import testing\n\nThis enables the g-api-r service in devstack, which allows tempest\nto run the remote import test, causing it to stage and import an\nimage across two different workers. Note we disable it for the\nstandalone mode, since devstack does not support starting another\nstandalone glance.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/770487\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/770520\nChange-Id: Ica715fc1922f4b36dd0bb008ef6706b86115ec05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/447602a703974a371f0f2d8e50dd12af8a12c9f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/447602a703974a371f0f2d8e50dd12af8a12c9f6"}]},"branch":"refs/heads/master"},"297d145529d7a174df96b5126b0dac2bfe0c7e43":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-03-01 18:56:27.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/29/770629/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/770629/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/770629/8"}}},"commit":{"parents":[{"commit":"1a4521a678263ddf859153530fb8c45c0390d7ec","subject":"Distributed image import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a4521a678263ddf859153530fb8c45c0390d7ec"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 15:41:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-01 18:56:24.000000000","tz":-480},"subject":"Enable second glance worker for import testing","message":"Enable second glance worker for import testing\n\nThis enables the g-api-r service in devstack, which allows tempest\nto run the remote import test, causing it to stage and import an\nimage across two different workers. Note we disable it for the\nstandalone mode, since devstack does not support starting another\nstandalone glance.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/770487\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/770520\nChange-Id: Ica715fc1922f4b36dd0bb008ef6706b86115ec05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/297d145529d7a174df96b5126b0dac2bfe0c7e43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/297d145529d7a174df96b5126b0dac2bfe0c7e43"}]},"branch":"refs/heads/master"},"cd49b3f4418db8e873466aef41bf96bb83aec0d7":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-03-01 18:59:15.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/29/770629/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/770629/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/770629/9"}}},"commit":{"parents":[{"commit":"dc77dd79a21df34720b40b427d43057c06e7cbac","subject":"Distributed image import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc77dd79a21df34720b40b427d43057c06e7cbac"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 15:41:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-01 18:59:08.000000000","tz":-480},"subject":"Enable second glance worker for import testing","message":"Enable second glance worker for import testing\n\nThis enables the g-api-r service in devstack, which allows tempest\nto run the remote import test, causing it to stage and import an\nimage across two different workers. Note we disable it for the\nstandalone mode, since devstack does not support starting another\nstandalone glance.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/770487\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/770520\nChange-Id: Ica715fc1922f4b36dd0bb008ef6706b86115ec05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cd49b3f4418db8e873466aef41bf96bb83aec0d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cd49b3f4418db8e873466aef41bf96bb83aec0d7"}]},"branch":"refs/heads/master"},"ac9630c79fd667d0c7d1fef5dfe710c57940ef59":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-03-02 19:53:18.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/29/770629/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/770629/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/770629/10"}}},"commit":{"parents":[{"commit":"41e1cecbe63c778ce8e92519993c61588ea1f0cb","subject":"Distributed image import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41e1cecbe63c778ce8e92519993c61588ea1f0cb"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 15:41:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-02 19:52:13.000000000","tz":-480},"subject":"Enable second glance worker for import testing","message":"Enable second glance worker for import testing\n\nThis enables the g-api-r service in devstack, which allows tempest\nto run the remote import test, causing it to stage and import an\nimage across two different workers. Note we disable it for the\nstandalone mode, since devstack does not support starting another\nstandalone glance.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/770487\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/770520\nChange-Id: Ica715fc1922f4b36dd0bb008ef6706b86115ec05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac9630c79fd667d0c7d1fef5dfe710c57940ef59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac9630c79fd667d0c7d1fef5dfe710c57940ef59"}]},"branch":"refs/heads/master"},"35cfac4c249346bae4e7fed2e9fbfc9b777f0e1d":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-03-03 14:37:56.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/29/770629/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/770629/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/770629/11"}}},"commit":{"parents":[{"commit":"d8a6309893d249623a31ecec88bb373fa08dfd3f","subject":"Add administrator docs for distributed-import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d8a6309893d249623a31ecec88bb373fa08dfd3f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 15:41:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-03 14:37:30.000000000","tz":-480},"subject":"Enable second glance worker for import testing","message":"Enable second glance worker for import testing\n\nThis enables the g-api-r service in devstack, which allows tempest\nto run the remote import test, causing it to stage and import an\nimage across two different workers. Note we disable it for the\nstandalone mode, since devstack does not support starting another\nstandalone glance.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/770487\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/770520\nChange-Id: Ica715fc1922f4b36dd0bb008ef6706b86115ec05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35cfac4c249346bae4e7fed2e9fbfc9b777f0e1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35cfac4c249346bae4e7fed2e9fbfc9b777f0e1d"}]},"branch":"refs/heads/master"},"d36488d09e94db6e67e1baac2a9f23c912863157":{"kind":"REWORK","_number":12,"created":"2021-03-03 23:18:32.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/29/770629/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/770629/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/770629/12"}}},"commit":{"parents":[{"commit":"d8a6309893d249623a31ecec88bb373fa08dfd3f","subject":"Add administrator docs for distributed-import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d8a6309893d249623a31ecec88bb373fa08dfd3f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 15:41:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-03 23:18:24.000000000","tz":-480},"subject":"Enable second glance worker for import testing","message":"Enable second glance worker for import testing\n\nThis enables the g-api-r service in devstack, which allows tempest\nto run the remote import test, causing it to stage and import an\nimage across two different workers. Note we disable it for the\nstandalone mode, since devstack does not support starting another\nstandalone glance.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/770487\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/770520\nChange-Id: Ica715fc1922f4b36dd0bb008ef6706b86115ec05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d36488d09e94db6e67e1baac2a9f23c912863157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d36488d09e94db6e67e1baac2a9f23c912863157"}]},"branch":"refs/heads/master"},"8f2cbc90b67b855d7d75ce9b9b929a0114448527":{"kind":"REWORK","_number":13,"created":"2021-03-04 16:48:36.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/29/770629/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/770629/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/770629/13"}}},"commit":{"parents":[{"commit":"fe8240562c4d8f673445079231fe3cff5848071b","subject":"Merge \"Add administrator docs for distributed-import\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fe8240562c4d8f673445079231fe3cff5848071b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 15:41:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 16:48:32.000000000","tz":-480},"subject":"Enable second glance worker for import testing","message":"Enable second glance worker for import testing\n\nThis enables the g-api-r service in devstack, which allows tempest\nto run the remote import test, causing it to stage and import an\nimage across two different workers. Note we disable it for the\nstandalone mode, since devstack does not support starting another\nstandalone glance.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/770487\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/770520\nChange-Id: Ica715fc1922f4b36dd0bb008ef6706b86115ec05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8f2cbc90b67b855d7d75ce9b9b929a0114448527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8f2cbc90b67b855d7d75ce9b9b929a0114448527"}]},"branch":"refs/heads/master"},"662607f496bf0267b265a0820a91783c15270ea4":{"kind":"REWORK","_number":14,"created":"2021-03-04 20:13:28.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/29/770629/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/770629/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/770629/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/770629/14"}}},"commit":{"parents":[{"commit":"fe8240562c4d8f673445079231fe3cff5848071b","subject":"Merge \"Add administrator docs for distributed-import\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fe8240562c4d8f673445079231fe3cff5848071b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-13 15:41:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 20:13:23.000000000","tz":-480},"subject":"Enable second glance worker for import testing","message":"Enable second glance worker for import testing\n\nThis enables the g-api-r service in devstack, which allows tempest\nto run the remote import test, causing it to stage and import an\nimage across two different workers. Note we disable it for the\nstandalone mode, since devstack does not support starting another\nstandalone glance.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/770487\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/770520\nChange-Id: Ica715fc1922f4b36dd0bb008ef6706b86115ec05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/662607f496bf0267b265a0820a91783c15270ea4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/662607f496bf0267b265a0820a91783c15270ea4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
