)]}'
{"id":"openstack%2Fglance~771070","triplet_id":"openstack%2Fglance~master~I31b4dae018d52ead773db25472013d783066ee17","project":"openstack/glance","branch":"master","topic":"os-glance-namespace","hashtags":[],"change_id":"I31b4dae018d52ead773db25472013d783066ee17","subject":"Make os_glance namespace reserved","status":"MERGED","created":"2021-01-15 19:47:00.000000000","updated":"2021-02-01 16:21:21.000000000","submitted":"2021-02-01 16:19:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"771070-os-glance-namespace","meta_rev_id":"47cf7b6be4bfd589daf3f3570515f2cdaaf627c2","_number":771070,"virtual_id_number":771070,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-01 16:19:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-02-01 16:21:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2021-02-01 16:21:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-02-01 14:45:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-15 19:50:05.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-01-16 02:04:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-16 02:13:48.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-22 13:49:26.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"4986075979c394acc060e82c658b1e103498d805","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-15 19:47:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bad381c974cc9c6ca95f13435d61d48a0d9cc7e3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-15 19:50:05.000000000","message":"Patch Set 1:\n\nTempest test: https://review.opendev.org/c/openstack/tempest/+/771071","accounts_in_message":[],"_revision_number":1},{"id":"96c412407119274a9c0254358a41aaba72922ae3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 02:04:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/23efe8049a5b41959b69203867302ef6 : SUCCESS in 57m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e90a0ac0e85b40c98cf41763a06ad784 : FAILURE in 1h 23m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/203e730519884dcdabf8bc6a913c43dc : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca4b0a804aea4ff880f38be1ff3e926b : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96126a03365240d5aaacd0d9e5dad5ab : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecfa0c1ac9ee4d24874db42c72c8a232 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32ddf1420b05484cbc278323d4d67afc : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58709d7ca410426ab83d1b46e32f969d : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0427f58bf8c044e9af538631439d08c9 : SUCCESS in 10m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/928c2b8e16164d7bb4e40f7aeabcf467 : SUCCESS in 7m 21s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2f91abb9b6934031b5a4e47f6c49e3ee : SUCCESS in 4m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3e779fa40d0442ce98f9fff86acd45d5 : FAILURE in 1h 34m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ea184354c48344989a398cddbcc889c3 : FAILURE in 1h 19m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b27e6f93059f4b63a9eae5a588628a96 : FAILURE in 1h 18m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/51f385c65ade4f3c8e8ff13fb76ce669 : FAILURE in 1h 16m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7321e8c9475741a4b620c7b3e2b929ef : SUCCESS in 55m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/989521d906194d1e848dd7da1a0d5fae : FAILURE in 1h 18m 40s","accounts_in_message":[],"_revision_number":1},{"id":"41f6664975d5034db3fb94b6dd5968750e1be098","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-16 02:13:48.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/771070/771070/1/check-not-vote/ext-glance-zuul/5e97bc6 : FAILURE in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"72c90d2c9f0a651d9771b7ab5c5527713fa3997e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-18 15:33:51.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5887961ffc36b49a424ce976709857465f37b92e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-01-18 15:54:10.000000000","message":"Patch Set 2: Code-Review+2\n\nThis makes sense.  As a followup, we should update docs and api-ref and add a release note.  I don\u0027t think it requires an API version bump because with property protections, it\u0027s been possible for a long time to include a \"restricted\" namespace specific to a particular cloud, so clients should be open to the possibility that a create-image request could be rejected due to specifying a restricted custom property.","accounts_in_message":[],"_revision_number":2},{"id":"6352cb505a044b4c6c9c6e34bd1ea3ef8e885b15","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-18 18:22:46.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/771070/771070/2/check-not-vote/ext-glance-zuul/93e0c09 : FAILURE in 8m 20s","accounts_in_message":[],"_revision_number":2},{"id":"b423485e1d169273a9d7e4e8317288abb6ce8898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 23:48:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f62fc56c1ece46b2b41b61dc2d5cd827 : SUCCESS in 1h 01m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e76d3eac7dc24d14b5912c6997240dab : SUCCESS in 1h 09m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fcce4b13b714aac8328999cc8b8fcf6 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e0a54b9cbfd4cd8bdb58cb8c144df7a : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a33798538f4e4bf4b8f96151b00310ca : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fece6f6bff8e401c9129e3cf06b9f42c : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ab73674f4ee41d5b8d5aa9c308d2918 : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d71bf09967d4a4890bd7872cacbd021 : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4e9f1ef4e87e4a70aedb5dbb22d5efab : SUCCESS in 8m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/29320f5d69d84c4985064753a3a83adf : SUCCESS in 7m 50s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ac86746aa98745388bd77f029ac18a30 : SUCCESS in 5m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/509da998993a4fefb5d345394493f8c1 : SUCCESS in 1h 09m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/72d86c36d9e645058bdef3e842dee1c6 : FAILURE in 1h 05m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9586e9dc74cb4d2db4ac7e6d59b488d1 : SUCCESS in 1h 44m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0088fbe5efc8491f9d94c6c36bb97ea3 : FAILURE in 55m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be55e3331f87482b8b312e57ac872253 : SUCCESS in 53m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f0e643ca925941718ef665042c292957 : FAILURE in 1h 56m 13s","accounts_in_message":[],"_revision_number":2},{"id":"d3887cacf2ce4bd670c0b2a4287c4107420278a3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-18 23:50:27.000000000","message":"Patch Set 2:\n\nrecheck notmyproblem","accounts_in_message":[],"_revision_number":2},{"id":"41b54e0360347da489b62e8ebd24518ea4013e34","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-18 23:55:30.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/771070/771070/2/check-not-vote/ext-glance-zuul/8b0c03c : FAILURE in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"ce958e8f8411ed3a933c67d1ef770b2988e227bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 02:44:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/474cfe05f78d45ffaa0c461abdfc5cac : SUCCESS in 1h 11m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/37ff9a4af81a426a971baab400a51784 : SUCCESS in 1h 26m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56befe47f23b4226a5d6df7fc69a7896 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18239b1246094145b1bce1135371a91f : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe48b9abac154db299ac52141ecf5eba : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e341b7e27504f77915de1649c52aa3a : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25506820b2614cf9bde184c24a332840 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b9482a9204f4f89a4c9b5cf8033fafb : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4f705e8151e046928b17aa13eb5277d0 : SUCCESS in 7m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7a11a2e2ac784824859e43941105bb44 : SUCCESS in 7m 51s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3628d1eeb3ac42e3b7f5e15b2fdb3419 : SUCCESS in 4m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/422b168d33ef4ec29182dffaf637fe59 : SUCCESS in 1h 17m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/392e49ba3ea04a1d967f463ce32bbeec : SUCCESS in 1h 15m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/20ec6873011a4e0d93f7f1432e1dfa34 : SUCCESS in 1h 18m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/498a21b91a0b4c6d866b9d1c18beb05e : SUCCESS in 1h 13m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6e7d5a8d4eb4c25be223909e3b489a1 : SUCCESS in 1h 04m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bae96e3a554c47eb932ff5e261cccaa9 : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":2},{"id":"c97bda94ae86bc72eacd99111ed090409fa81c0a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-01-22 13:49:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)\n\nnit inside but can be addressed later","accounts_in_message":[],"_revision_number":2},{"id":"7a8608eb40cf11f41c8b1c2541008df181e7a320","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 17:03:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d603a9c12f8e98dd8b127884752665f3dd224a5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 19:00:32.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"935bb30e050812f219139aec716c495888dc33b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-22 21:02:48.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"77f92d553daf10bd16f5f3053edcc479987180d9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-22 22:34:24.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/771070/771070/3/check-not-vote/ext-glance-zuul/1357fda : FAILURE in 4m 50s","accounts_in_message":[],"_revision_number":3},{"id":"6643112cf976a90383a0e0be8747cdce44b3f4db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 23:13:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5c7ddd161449449da6f60f71229ab6b7 : SUCCESS in 56m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ddab7943126544a3890fe2be6aef7dc6 : SUCCESS in 1h 19m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47c4ca230a294eae8879c707fa75db85 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3365bc4297c348c6a2310df6937ae892 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d8ccd8274d0462080493fed79fc3cd7 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eecf0e92214d48c99fd5befa19889bd6 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96afd7667f66497298e784406246699d : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fead7008dd24bea94727686268660ab : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/078bd7ed49b5448f9b5730226d103ed6 : SUCCESS in 8m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3ffa10063ef542a980fa76683e4f6f43 : SUCCESS in 7m 18s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ad7c5c224d454c61981419e68dd7621f : FAILURE in 5m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/408018b9c58441e38885243e483c1015 : SUCCESS in 1h 16m 04s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a0b714a5c6c6494f887f97d3267d45d4 : SUCCESS in 1h 17m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6b9f1481b32f4b4f821fb1e2be8b3396 : FAILURE in 1h 05m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7e09b435b1fe4ada86b4830fd50add46 : SUCCESS in 1h 26m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76fa1380b7274b1eac50d3d0f5aa5f16 : SUCCESS in 57m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7f569d85e8574febaf9491d03a8b87eb : SUCCESS in 1h 26m 39s","accounts_in_message":[],"_revision_number":3},{"id":"ac181832c45ad0e9d1b78917ea8950fe28e50546","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-25 19:33:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"5539f9409a34b1e6d7385a54bc14d0d1bf969bb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 22:44:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ca8d68f0843e47fa8cf263982c4a2272 : SUCCESS in 1h 04m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a6c2e982ef184f83a82620bb6e13500a : SUCCESS in 1h 20m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52905e8cd7f443d7a34e00b0b7a0fa61 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7fa76da8a994b50b912774b3f84d168 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c525ba7ec2d14485b1b4d8029dd116d6 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/668c5bbe03b74bceb3136ed6512fe334 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2392346a2bb458e8b3cace1e87c0d10 : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58fbd0626b65434c90404b94790109c3 : SUCCESS in 10m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a6687a511414bd48c589806f99be735 : SUCCESS in 7m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4046f4270022460198e06f6ac7b0e996 : SUCCESS in 8m 21s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/117464e257fc46269523371b02dfaaa8 : SUCCESS in 10m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/10b4a26c3b324e67939d1f0f07a1f90c : SUCCESS in 1h 20m 47s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/aff980f584b74d018860bb949fcee8f0 : SUCCESS in 1h 14m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d088004e8d2f4f74ac903ba5c0ea1763 : FAILURE in 1h 01m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ebfc852178c84494b27cba6947f300d2 : FAILURE in 1h 16m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6e3097ebff3461fbb034efccc0e3af9 : SUCCESS in 1h 10m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/faddf8bb0764458a9fe65ca19f2116f6 : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":4},{"id":"d2ec4ccbdb85e8681b163ae6331978a7ae84bd99","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-27 16:57:35.000000000","message":"Patch Set 4:\n\nrecheck cinder fail","accounts_in_message":[],"_revision_number":4},{"id":"ba2d444828d4cf1477f5d9a340e9bb93ba4020c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 22:02:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e3592a0281864e86b21d9ef9450de9b9 : SUCCESS in 1h 00m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/658d118049504397a315c803dfb7d23c : SUCCESS in 1h 11m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de85ea70a378413ea2c8ede8aa56a607 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cd6326a62344e00afee16e042010d29 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/327533f46a074814a3bbc09e3fa4235e : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c71051de1de45a8b1b0ec1f0f24f777 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80baa88ecc234a74bf149d61ad9cb461 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83f2f382d7674f83ae59d86f44187672 : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65c65f5b05d542c39f8bbe0a90462e3a : SUCCESS in 8m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/49930d1a2f2b42e7aab3f6fca15828f1 : SUCCESS in 7m 34s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4aefa4e837ff45579c2362669466e081 : SUCCESS in 6m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1bd0963d33a04b149005cf64f219cff6 : SUCCESS in 1h 11m 29s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/07de8335259e4ac59952804fe196d1cb : SUCCESS in 1h 12m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9d555688ed2f4bf9a52db32a46f8b900 : SUCCESS in 1h 15m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/72ffb09807794b83a84ecb11c110f666 : SUCCESS in 1h 33m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c27e9d2096440e0b94855276396ac00 : SUCCESS in 1h 00m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/29f72e938da0462597c5ac54e950f505 : SUCCESS in 1h 12m 15s","accounts_in_message":[],"_revision_number":4},{"id":"f825c34eef2f6486a79dbab83db03df859a7af57","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-01 14:45:08.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"749005c1c5a8c5f7e45b0f398cf793d74bb193a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 14:45:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7906f693fe01fcbd7f1e8bb456eab3d6a2809ade","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 16:19:26.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/81109fc41216430d811425b3d2de5064 : SUCCESS in 56m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/37b76482c9cd4c0daddf6933247f9b93 : SUCCESS in 1h 07m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/973aafb46f1546a79f4233164d2ed602 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16dd367842954d0bbec8ca1391620b43 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43a246adb1e740e5a775374d51677644 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ee9d8817adc40d2a4d7dfe214151ac3 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef1df922f53443d965415948415dcca : SUCCESS in 8m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5cf3c3a3d5574bf5998eb5fa7eaaf684 : SUCCESS in 7m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c4e605e4cf9b4f7ba3c8d00757dcc351 : SUCCESS in 8m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/57d13227d5f04753b022b8636c142427 : SUCCESS in 1h 26m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d7a78d226ad34dca87649496c552988e : SUCCESS in 1h 17m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b602c5f9b824cdfba3e7a8c067e8084 : SUCCESS in 55m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c4867f2293624a4d924c4339cf329e8c : SUCCESS in 1h 19m 49s","accounts_in_message":[],"_revision_number":4},{"id":"0a7c0d2aa6be0379c4c9d9889c8a2e48e0961c6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 16:19:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"4b69de38c9698d15240d492f4a0b05ad5b82a884","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 16:21:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/536e70cd3a0540599311d5f9f9a1794e : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0c45de3ed8253096bb9d4031db8e046107e2774b","revisions":{"5a5c142ab47f6825eff8d8318d88b97a2091a841":{"kind":"REWORK","_number":1,"created":"2021-01-15 19:47:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/70/771070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/771070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/771070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/771070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/771070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/771070/1"}}},"commit":{"parents":[{"commit":"73742be99944d923031aa5f90e06051126b17007","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/73742be99944d923031aa5f90e06051126b17007"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-15 20:38:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-15 20:38:35.000000000","tz":-480},"subject":"Make os_glance namespace reserved","message":"Make os_glance namespace reserved\n\nThis adds a general mechanism for reserving property names that start\nwith os_glance. This has been done informally already, but no\nenforcement was performed, except for specific keys on update. As a\nresult, banning these keys from create, for example, was missed and\nusers are able to set these keys during an POST /images operation.\n\nChange-Id: I31b4dae018d52ead773db25472013d783066ee17\nCloses-Bug: #1912001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5a5c142ab47f6825eff8d8318d88b97a2091a841"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5a5c142ab47f6825eff8d8318d88b97a2091a841"}]},"branch":"refs/heads/master"},"1a779888dd3e68b26d8b54bd1ddf15c3284cc462":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-18 15:33:51.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/70/771070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/771070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/771070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/771070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/771070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/771070/2"}}},"commit":{"parents":[{"commit":"73742be99944d923031aa5f90e06051126b17007","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/73742be99944d923031aa5f90e06051126b17007"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-15 20:38:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-18 16:29:54.000000000","tz":-480},"subject":"Make os_glance namespace reserved","message":"Make os_glance namespace reserved\n\nThis adds a general mechanism for reserving property names that start\nwith os_glance. This has been done informally already, but no\nenforcement was performed, except for specific keys on update. As a\nresult, banning these keys from create, for example, was missed and\nusers are able to set these keys during an POST /images operation.\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/771234\nChange-Id: I31b4dae018d52ead773db25472013d783066ee17\nCloses-Bug: #1912001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a779888dd3e68b26d8b54bd1ddf15c3284cc462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a779888dd3e68b26d8b54bd1ddf15c3284cc462"}]},"branch":"refs/heads/master"},"1800be495494bd422ae418f7a3631ac02d2f9954":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-22 21:02:48.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/70/771070/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/771070/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/771070/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/771070/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/771070/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/771070/3"}}},"commit":{"parents":[{"commit":"cbc8519554362d49539e22dfe394c43df0448e11","subject":"Merge \"Fix a typo in functional helper class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cbc8519554362d49539e22dfe394c43df0448e11"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-15 20:38:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-22 22:00:27.000000000","tz":-480},"subject":"Make os_glance namespace reserved","message":"Make os_glance namespace reserved\n\nThis adds a general mechanism for reserving property names that start\nwith os_glance. This has been done informally already, but no\nenforcement was performed, except for specific keys on update. As a\nresult, banning these keys from create, for example, was missed and\nusers are able to set these keys during an POST /images operation.\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/771234\nChange-Id: I31b4dae018d52ead773db25472013d783066ee17\nCloses-Bug: #1912001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1800be495494bd422ae418f7a3631ac02d2f9954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1800be495494bd422ae418f7a3631ac02d2f9954"}]},"branch":"refs/heads/master"},"0c45de3ed8253096bb9d4031db8e046107e2774b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-01-25 19:33:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/70/771070/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/771070/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/771070/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/771070/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/771070/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/771070/4"}}},"commit":{"parents":[{"commit":"03bf00ee3b4c95713795f6bb7fcb4f37e98b5866","subject":"Change database migration version to wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03bf00ee3b4c95713795f6bb7fcb4f37e98b5866"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-15 20:38:35.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-25 20:30:50.000000000","tz":-480},"subject":"Make os_glance namespace reserved","message":"Make os_glance namespace reserved\n\nThis adds a general mechanism for reserving property names that start\nwith os_glance. This has been done informally already, but no\nenforcement was performed, except for specific keys on update. As a\nresult, banning these keys from create, for example, was missed and\nusers are able to set these keys during an POST /images operation.\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/771234\nChange-Id: I31b4dae018d52ead773db25472013d783066ee17\nCloses-Bug: #1912001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c45de3ed8253096bb9d4031db8e046107e2774b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c45de3ed8253096bb9d4031db8e046107e2774b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
