)]}'
{"id":"openstack%2Fglance~771907","triplet_id":"openstack%2Fglance~master~I5c57696827719d752879cb11044d9e969c853411","project":"openstack/glance","branch":"master","topic":"os-glance-namespace","hashtags":[],"change_id":"I5c57696827719d752879cb11044d9e969c853411","subject":"Change database migration version to wallaby","status":"MERGED","created":"2021-01-22 04:53:30.000000000","updated":"2021-01-28 00:18:56.000000000","submitted":"2021-01-28 00:17:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"771907-os-glance-namespace","meta_rev_id":"592aac6d87c7f319118ac471725835f396b28ef4","_number":771907,"virtual_id_number":771907,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-28 00:17:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2021-01-27 14:23:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2021-01-28 00:18:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2021-01-27 14:23:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-22 05:45:11.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-22 06:38:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-22 06:48:16.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-01-22 06:48:16.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-01-22 06:48:16.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-01-22 06:48:16.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"abb2b8e46f1b85ca61e8c3f4cbe024e776c26e59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-01-22 04:53:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62bea937202d86af19d573eac7c28ab1c3e6ca1d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-22 05:45:11.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/771907/771907/1/check-not-vote/ext-glance-zuul/04b2556 : FAILURE in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d041bc27ddf12bdee807a117a64bb36e7f739862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 06:38:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/0abd335ea93f4b7589e4749f5f6976fe : SUCCESS in 1h 00m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/072e4bd7e61d4ee4b5ab40812f698c5e : FAILURE in 1h 43m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e426169003104088a5bf51dd32803613 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2baf74f0aa114d3da9b0b578bfac9bfc : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ab171104d0444a19f052f9647cb4e8d : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b059c8ef2fbb43e4bc280b78d911d0df : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f9133374b98409f9a0bf8ebcafdb3de : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b50ca9d1b12b49c38646aa1ee213f9de : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/56ebda68d02a4298861b4b78e4f465e0 : SUCCESS in 6m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bce5c21534934b5ead7ed187f20175ac : SUCCESS in 6m 50s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c63a56b583fc481aa8329aa9460ba706 : SUCCESS in 5m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f9d779242c894a6da4a62a3f53c7153a : SUCCESS in 1h 38m 52s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f59cb7d554d74836828356c8f6c43c3d : FAILURE in 1h 11m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d128d96dec014054bc230d8757b082b8 : POST_FAILURE in 47m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7d539eff9d02468fb8188966af11448f : FAILURE in 1h 42m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2117949508444c24a344f59763bbffb0 : FAILURE in 48m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e8d2de3c405440f7a18028a3cf1c99bf : SUCCESS in 1h 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"54c1ad2e5bb62a310b4975fb8e51ee2cdb67c72c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-01-22 06:47:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c4c6920f059f2fd6a8744cf4e1b3de46d5e41110","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-22 06:53:07.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/771907/771907/1/check-not-vote/ext-glance-zuul/009cff0 : FAILURE in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"43c014f39487f951e94f502cbb80a6e83eb6ebb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 08:31:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/bee3531854664787948f6e02c33bf1d5 : SUCCESS in 50m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/11d9e762676e47b1adcd6cf4aec65289 : FAILURE in 1h 07m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fd6390d2b5f4ce18b1c20df245743c0 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78cc9eb9a3774f69ac924db46f48fca4 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30d93758385e4380b4450c7e84d31046 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9afbee3efdc942bc9c42e685142bab2e : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27489a76ca2f4bb081e497fbfc15187f : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22b7973c70684563835f80c96df19e1d : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b6f1642d0d904bd8ae41b97ae162455c : SUCCESS in 8m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/735b3319ef6c46929294cb1b44a8868a : SUCCESS in 7m 24s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/40b609e3ebb443e690e1591f9ef9f1d5 : SUCCESS in 4m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/97f67918732c437891e1882c6f1a3da9 : SUCCESS in 1h 03m 11s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9c60bfb6eef24e0ea9dfb3b0de83c523 : SUCCESS in 1h 16m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4d29839bf2404d68b9d8f4cf911e9ac0 : SUCCESS in 1h 13m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/40e1072d0ca242e186dc40bd3b4757b7 : FAILURE in 1h 27m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b822964f6dc04986b4a43834e431e0fa : SUCCESS in 55m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0a0cd54caa1546c2813c992a6e291f44 : SUCCESS in 1h 07m 32s","accounts_in_message":[],"_revision_number":1},{"id":"24878b5d6be4570e78d72a76dd9589cdf074989f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-01-22 08:33:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c3daf33a8604d15a1a6d9958fd871656cb116aa2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-22 08:39:19.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/771907/771907/1/check-not-vote/ext-glance-zuul/51303bc : FAILURE in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a0e15a065cc8851bfc98f427357d90bae1d20896","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 10:30:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5581c14064f44d189d2a3182562ff84c : SUCCESS in 1h 08m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/81a1525699c24c81883a513d1046b008 : FAILURE in 1h 44m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f054a8cf9414f199e76b0383ce402c6 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a25c41f8be4bdfbc4026103f097ff0 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/576ce0385d2c43aa8a48a4a526de557c : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a96c939726d471fbfbc86e546fd0720 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bd28bbd19d448fea80cd902c18fe9c1 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f7acb7020640c598140ddbf5bb1c8a : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4e1928449ec84c259c6f0ca2d43d01a1 : SUCCESS in 8m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a03bb298061442b6b7489ba983bfe5c5 : SUCCESS in 9m 32s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c82e9231d9034a78b7028b3e59eb8c8c : SUCCESS in 4m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d7f20763be504d0e992e5146ad57b6de : SUCCESS in 1h 18m 11s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bc1ac46718794793aa0fd825547dbae9 : SUCCESS in 1h 21m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/36fe077f4f26432e8f5226b8f08e9d61 : SUCCESS in 1h 55m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/37575f083fb447f9805705a0635ccab9 : SUCCESS in 1h 19m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e875c1fa8e0a4b86af5c817156b648c2 : SUCCESS in 52m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c5c0ef651f3c42478b1539a992b67eec : SUCCESS in 1h 08m 19s","accounts_in_message":[],"_revision_number":1},{"id":"38c54d4f4726217fcbc4ba819643d3ee1a5151af","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-01-22 13:41:52.000000000","message":"Patch Set 1:\n\nrecheck unrelated failure","accounts_in_message":[],"_revision_number":1},{"id":"4766fbb087b637d42d2cf7e723ac278b3ce56882","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-22 13:47:22.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/771907/771907/1/check-not-vote/ext-glance-zuul/11f2f5d : FAILURE in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"da5d6417f4d8f79ba70a1d9eae23aad38c50feed","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-22 14:43:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9d86bd7aa80dd9d83e72d511389e1ff22fb10bd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-01-22 14:59:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c3609d49f49719bb590978c35733e1bcdb4b573","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-22 15:46:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"62e62badcd53e9999521dfc16ec5378baa25121f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 16:05:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/70536672555846c9bdf01f026bc0b0a2 : SUCCESS in 1h 01m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b051df48faa54be7aaa31f0d49197b41 : SUCCESS in 1h 34m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0712815241e94cb2adfcd3fa409b88bc : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb5c537ad5a4f1d9d3216eb1adeb317 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ccbe95b17e7a4da6a99f2d02dd29ee21 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca1530e2e7f748c58f82f108d6b2f5e2 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5870112626d64b538a35ef142ca6fe09 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e623d7c516f8451997161c2edd8df430 : SUCCESS in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a72c2a292940486b9d8693110848cb13 : SUCCESS in 8m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4df06a801f584690a4df4bff6757ade1 : SUCCESS in 9m 38s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7ad2459873e242a78b5eddc70c22cdf0 : SUCCESS in 6m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8220e62ef74d4e879232cfbc737dd130 : SUCCESS in 1h 17m 05s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b65e396a9a2b46b3a47c880461a1f640 : SUCCESS in 1h 58m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/84fb3bb96af04e908b6450cdb9b013fb : SUCCESS in 1h 14m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b7ebf9fca7fc45bd965ab7ff35bea53a : SUCCESS in 1h 12m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f0628cf7d60406ba244879e30b76734 : SUCCESS in 1h 05m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/23a5a70a3edb47d58daddb2d630aed3f : SUCCESS in 1h 03m 58s","accounts_in_message":[],"_revision_number":1},{"id":"793c8f9d57e80f1b837219c75880f9a8e1f1161d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-25 19:33:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"25968b6007fad26855a791b8208def6f0337419b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 23:02:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/97faaec434ec43dcaf4713224294ccfb : SUCCESS in 59m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a8af12743e54d529d542bd45330ba5e : SUCCESS in 1h 15m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dee49da0378b4037b1eb87785910dd7a : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/734221434dfd4fa7b74bda07b12cbb35 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a82828bc8bd443218384af832f880201 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c0e21aa331742aa884cf8940734ff33 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2734f301363940ca9863cc14fe26d890 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b453c8b42b46b3b0feb9af79d12c54 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b685149d82be438bb5889e96d243cc16 : SUCCESS in 7m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b00acf36f3354f4fad1db950d829588d : SUCCESS in 8m 39s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ad701c0c7cef42128f6ea385d988f227 : SUCCESS in 4m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/91f560becf7c46d68a247c479892ed84 : SUCCESS in 1h 07m 00s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f72e8709ff334e8f885be5e2bfa5a91d : SUCCESS in 1h 06m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/46ce57ce9c924c4d93d34881c61bcddc : SUCCESS in 1h 44m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/20198a8302144ae489c67372f2d1b911 : SUCCESS in 1h 19m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a7b79ba59ca4decb6694fd8cb4501e7 : SUCCESS in 1h 02m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4602d82be4e74f288a2830437db81691 : SUCCESS in 1h 05m 54s","accounts_in_message":[],"_revision_number":2},{"id":"769e030b10e1f8bdbe78d1736be08319c2527c24","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-01-27 14:23:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8cca4914354580b84745eb8a150aaa561d646d76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 14:23:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"03e538c28d1eddc66da3df64bcdbb304fe71ab5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 17:14:33.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/98fb614a8c824f7fab0daebd80f91771 : SUCCESS in 58m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d13f003b35e24119b2a24ea691e7004e : SUCCESS in 1h 30m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcfcfe649688400f92dc365951fedd9b : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecd6769dcc1e408ea80876520915a254 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05ca6f98f4f149a299b4b55a65d35cbb : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7b04086d4c24cb38698d2fffa2e1681 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cbd20160fc14e9aa5ef7fa3d243806c : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/89bfd7645b14401d85f741084c5e6a32 : SUCCESS in 8m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ebc4572162b64199b2efd49ee7e5470a : SUCCESS in 7m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f2ba5f00db174363839db6d949c28585 : SUCCESS in 1h 21m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8ab0c3263dcc468cb962520427e3d4bc : FAILURE in 1h 21m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64919f4e09d04cc591d0d2ded79d3a5b : SUCCESS in 1h 09m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dbbbda321c77487aa092300d1cd2a483 : SUCCESS in 1h 25m 43s","accounts_in_message":[],"_revision_number":2},{"id":"6ef2671aff416d9d7fc55cc6f168d46bc44951ad","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-27 17:30:45.000000000","message":"Patch Set 2:\n\nrecheck cinder fail again","accounts_in_message":[],"_revision_number":2},{"id":"b06edf2bc2c5ba507cb3a66ffe9d9709f66cf460","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-27 20:44:33.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/771907/771907/2/check-not-vote/ext-glance-zuul/a8c8c58 : FAILURE in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"946106db0a86833380efeb96b3a6ffb93f0fa5d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 22:41:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5a8e42f10d074118b8762849998a4453 : SUCCESS in 1h 03m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d3fed58611142198b111b7f591b75fa : SUCCESS in 1h 13m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f07c2a582bad42ef9c3850bf8f47ed91 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/726a3948bab64a94b4bee0c76217a77e : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d161b68d327c4192bb92ed8f63fc43c7 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65d04c38576f42cd997e814117cb17ba : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b7f27a93adc482489929473c6b8f358 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/844bf81c94f24b1f800e46f750c4057b : SUCCESS in 7m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5fb517d4354049a08b397ed25708cc7c : SUCCESS in 7m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b0c708e8d5b44b738b4075d593df59db : SUCCESS in 7m 39s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f9997056042040ff816895ea84e49f91 : SUCCESS in 4m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6fb2bbb153f947d3bc16ccc2cae6f5f4 : SUCCESS in 1h 14m 24s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8fb20ceeec1b4f6689907ee36a4b5edb : SUCCESS in 1h 13m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bd35e343889a4bd48a245277567b8c14 : SUCCESS in 1h 03m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/125d76cc0eda4a479009ce7a9cc4bef1 : SUCCESS in 1h 33m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56e88278955a44f6abb11bc3fb1a0fb6 : SUCCESS in 1h 11m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2fa58441d3d3453fbbe978746bc5398d : SUCCESS in 1h 38m 07s","accounts_in_message":[],"_revision_number":2},{"id":"79f1f35072122cb6afeed68fb6e80744c4dec4cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 22:41:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c7c5d147842109aa92566abbaea1907e8139121d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 00:17:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/33b42304dcd34ffc9673fbd28fd66964 : SUCCESS in 1h 09m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b001b649fff34ed882bae44d6eea1f31 : SUCCESS in 1h 11m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8bf44c9822d41e0a15a5c6aa904c856 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/067e2813ba7c4edb836eca9998da0ff0 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8ce8f17a8fe47dba58b67e3bba95b02 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a0352a316a14029a0f8d3857adf6ffe : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c7e12f84854599b5313d3ae6368fd0 : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a9d70d732c664ecc92fe9ad217682d4d : SUCCESS in 8m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fa8c6cf671d44849814e55664bd42b25 : SUCCESS in 7m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/514e1e629e534f5ca6d3e90c54fab80a : SUCCESS in 1h 17m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/690829647f594c6985b286ee728be822 : SUCCESS in 1h 09m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d0a3f6be99a40068d20d5ee448931fe : SUCCESS in 1h 02m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/780e583efb2b416b81eb5435d148d7f2 : SUCCESS in 1h 29m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6886eb2e2e11e5be69673c912032cbf5db80b9ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 00:17:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1618264dc0e0a505d2fc874c976eba189b5d2b69","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 00:18:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b71cdc0ea52f47aeb7515e7af780e138 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"03bf00ee3b4c95713795f6bb7fcb4f37e98b5866","revisions":{"ebdf2f1e529f967559ecafed677c535266014f28":{"kind":"REWORK","_number":1,"created":"2021-01-22 04:53:30.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/07/771907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/771907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/771907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/771907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/771907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/771907/1"}}},"commit":{"parents":[{"commit":"a7f0e44699a4c74c631526664dbeb970601c003d","subject":"Merge \"Fix upgrade checks for sheepdog\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7f0e44699a4c74c631526664dbeb970601c003d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-01-22 04:47:06.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-01-22 04:47:06.000000000","tz":0},"subject":"Change database migration version to wallaby","message":"Change database migration version to wallaby\n\nChange-Id: I5c57696827719d752879cb11044d9e969c853411\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebdf2f1e529f967559ecafed677c535266014f28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebdf2f1e529f967559ecafed677c535266014f28"}]},"branch":"refs/heads/master"},"03bf00ee3b4c95713795f6bb7fcb4f37e98b5866":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-25 19:33:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/07/771907/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/771907/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/771907/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/771907/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/771907/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/771907/2"}}},"commit":{"parents":[{"commit":"cbc8519554362d49539e22dfe394c43df0448e11","subject":"Merge \"Fix a typo in functional helper class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cbc8519554362d49539e22dfe394c43df0448e11"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-01-22 04:47:06.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-01-25 20:30:50.000000000","tz":-480},"subject":"Change database migration version to wallaby","message":"Change database migration version to wallaby\n\nChange-Id: I5c57696827719d752879cb11044d9e969c853411\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03bf00ee3b4c95713795f6bb7fcb4f37e98b5866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03bf00ee3b4c95713795f6bb7fcb4f37e98b5866"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
