)]}'
{"id":"openstack%2Fglance~773838","triplet_id":"openstack%2Fglance~master~I92f3a35ae5d32c5204fa9a8a6d9e4d5664ab6c8a","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"I92f3a35ae5d32c5204fa9a8a6d9e4d5664ab6c8a","subject":"[DNM] Test Revert \"Remove native ssl support\"","status":"ABANDONED","created":"2021-02-03 06:42:28.000000000","updated":"2021-03-04 12:10:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1090108a2f0aa9e0532bdf7f031c3908342f6f7e","_number":773838,"virtual_id_number":773838,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-03 08:02:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-03 06:48:25.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-03 08:02:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"856855e4fc5f239b72d0738fb9ba5914e1e60a3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-02-03 06:42:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b69c96237e9c2b8eac7f1a488835375d5a1f4525","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-03 06:48:25.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/773838/773838/1/check-not-vote/ext-glance-zuul/27fa398 : FAILURE in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"fa7bbab23c8032fed4c93c649073de9033a0bad2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 08:02:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d440210c83dc42e9b293a275366cdaaa : SUCCESS in 4m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/add29cb47d074f2f92bfe33bbab0d159 : SUCCESS in 8m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/590006c7ff79416cb2a8684a119a37fe : SUCCESS in 7m 33s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/278b48c143c440f6ad83ac35146e7878 : SUCCESS in 7m 24s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/0aa9fbba578d4c169e390473e555eb8f : FAILURE in 6m 46s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/10f47d0405ad4b6fae4106414be5671a : SUCCESS in 4m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/42789ed890c14edb8aaa047ca28fdf16 : SUCCESS in 1h 03m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1090108a2f0aa9e0532bdf7f031c3908342f6f7e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-03-04 12:10:36.000000000","message":"Abandoned\n\nDiscussed with Abhishek, it will not be included atleast in wallaby, he will be discussing in PTG for next releases. On our side we are going with enable ssl in all except glance https://review.opendev.org/c/openstack/puppet-openstack-integration/+/773336","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"891ee8598bcd01f72fc092a221ff3a30802d46e6","revisions":{"891ee8598bcd01f72fc092a221ff3a30802d46e6":{"kind":"REWORK","_number":1,"created":"2021-02-03 06:42:28.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/38/773838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/38/773838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/773838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/773838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/38/773838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/38/773838/1"}}},"commit":{"parents":[{"commit":"5eca0f66dbd26336f5c6366988191dbfa5e24c26","subject":"Merge \"Add policy sample file in doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5eca0f66dbd26336f5c6366988191dbfa5e24c26"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-02-03 05:34:29.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-02-03 06:41:52.000000000","tz":330},"subject":"[DNM] Test Revert \"Remove native ssl support\"","message":"[DNM] Test Revert \"Remove native ssl support\"\n\nSee if it works with recent eventlet, in some of\nour tests it seems to work now, trying it in gate.\n\nThis reverts commit 06b2465f598c014451c244d47c2ded4f1e2c6947.\n\nChange-Id: I92f3a35ae5d32c5204fa9a8a6d9e4d5664ab6c8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/891ee8598bcd01f72fc092a221ff3a30802d46e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/891ee8598bcd01f72fc092a221ff3a30802d46e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
