)]}'
{"id":"openstack%2Fglance~774308","triplet_id":"openstack%2Fglance~master~I90d97c7181bd2d449615732a61bd3d8074e93adc","project":"openstack/glance","branch":"master","topic":"secure-rbac-handle-members","hashtags":[],"change_id":"I90d97c7181bd2d449615732a61bd3d8074e93adc","subject":"Explicitly unset tempest_roles in devstack plugin","status":"ABANDONED","created":"2021-02-05 21:15:14.000000000","updated":"2021-02-23 22:23:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8621369775adbeffe7e93dca1db421681e934f3f","_number":774308,"virtual_id_number":774308,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-02-23 18:18:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-02-23 18:18:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-02-23 22:23:53.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-05 23:01:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-12 05:00:05.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-02-16 07:07:50.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-02-19 14:47:08.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-02-23 15:28:56.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"7c944bb7808eca763fecf93480c44bd6070d2368","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-05 21:15:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae88fb607a947f2606b04f4650ff47318c90790b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 23:01:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f82a3dc43a6e4afc87a1bc38581b1293 : SUCCESS in 57m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/432e12c80f5d40baa1fa0936aa33963f : SUCCESS in 1h 08m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26dcd54e620842168e0853b61745c8a3 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44e28bef5f49477fb5ac063714123fe5 : FAILURE in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb9f2b5b2bf642cda4f33ffdd2d096a6 : FAILURE in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd1f4f5cac0840fab13064c966d0a9e3 : FAILURE in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a20719fdfd27460690806faf70b28d31 : FAILURE in 3m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e7394039564a0a8a1e997ac2458d3d : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d381e8165d014e0f869efdcf6641f268 : FAILURE in 4m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cb289832830b448cb0930ceabd31bbbe : FAILURE in 3m 22s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8c0b1b224c0c4e82b975e705c103856f : FAILURE in 3m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/159abf737acb42809d12a0804c8c2937 : SUCCESS in 1h 06m 57s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5047a9d337684816997222dfac7f1301 : SUCCESS in 1h 20m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/91fa00b9a06e4c499a20b52febd93828 : SUCCESS in 1h 12m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/116b44396c8c48d697a8e3708857fc63 : SUCCESS in 1h 42m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e66d5e364924ae0896d165a96488fac : SUCCESS in 52m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6df0e1f4ad3b489587256dd334c66040 : SUCCESS in 1h 25m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1d12d943737bf53c9df204c9668004fe1a7a20e2","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-07 02:35:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"421710e0957b465709ea9add9986982bdc11aa0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 04:19:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/29331155bea547608d803e7644db3e87 : SUCCESS in 1h 03m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/452a82b1281f419e8e1d7585b1c255bb : SUCCESS in 1h 41m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2875dbd762c3431fa77a4b9ba9cf4cc4 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8493f392d1541c88c327a64476ef7cd : FAILURE in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a9e46cc28824e43aa5d47b67de68c58 : FAILURE in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2460259875df429998b8dca2188b2378 : FAILURE in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee471ee253b94f2180bddf5e13e0ad7b : FAILURE in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28efaba88dbc4ab880d706a2c0ce2d2d : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7aa0c8d6c6c446669c7d64c952c5a9cb : FAILURE in 4m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/10a02ae1fc6c44888e100b52a820224b : FAILURE in 3m 07s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8f787086211c4c548809cf06950aabe2 : FAILURE in 3m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e66465756db44c33bc540a72ee973111 : SUCCESS in 1h 38m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a5366547fc2e4b74bae427f5200dcda7 : SUCCESS in 1h 13m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/186b8c4eb13e4328b55ddd33c0dc3592 : SUCCESS in 1h 15m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8b4e056efff9468aab9df8d77617dcc8 : SUCCESS in 1h 32m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af516df62a1c47adadf0ded534851f5a : SUCCESS in 1h 12m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/611e038834f748b08e55a5029e1c3674 : SUCCESS in 1h 15m 18s","accounts_in_message":[],"_revision_number":1},{"id":"392fde36608d9a92680ee7bb7d2224f7a89f08cd","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-08 14:14:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0cd80db4e1f5a28497c0dc8de1649c075775f99f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 15:55:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/02dd8dd9aa1c4c478b69400c51153a93 : SUCCESS in 1h 10m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b1c340c79f541c3988048e4f3066903 : SUCCESS in 1h 15m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0475b39a76845b986e213870be2379d : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74fc903e04274b96992c19c7c38fde4f : FAILURE in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4873916d7ece436fa36b3a286b53c4da : FAILURE in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7577e66c514144df8442c51f42b072d4 : FAILURE in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ba2c7eb3984493fa632e14b8f12016b : FAILURE in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b111a62ff2c48679e3ece30f5536928 : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef6d4df4376c439b92c8dffe7826727d : FAILURE in 3m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8f091e487a5849369b250d695ced2cc4 : FAILURE in 3m 57s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/772b9970713d40c2ada847a2e8ad35c4 : FAILURE in 4m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b0c563409c024996b39a6cd3af7fea89 : SUCCESS in 1h 03m 50s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f73957ba32d04b3f9b0b8920275f2585 : SUCCESS in 1h 19m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/02147f93a8ba4fd2938ec660b67c2f6b : SUCCESS in 1h 21m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/63d3e15e14bb466f9efd6040e4df78f2 : SUCCESS in 1h 16m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2c879935f924f90b4616dab12f00cce : SUCCESS in 53m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/89e83b50add24a7da293b244cc198d87 : SUCCESS in 1h 15m 51s","accounts_in_message":[],"_revision_number":1},{"id":"dd40a3a5ceec7e56927af9553b86bac3b28bb5a8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-10 23:08:49.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"83aa9d041120b61da324125e487631a9a47f0b41","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-11 23:55:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bd854d62c1a6e093ff24d744a78666e6d0c2341b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 02:17:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a279df9f515647eaacfc0ab141cc27e3 : SUCCESS in 1h 00m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f272527b9c334543a11b1f8c8c16e572 : SUCCESS in 1h 18m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc121293b60b4705882e8ec8ebf30dc6 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14248e8e667b478e9ecd3564393b00d3 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5323ad2c4f1b4df1b4b39476746a7bab : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b710e177f844854bebdfec62e7b08a9 : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/932ec7d93d9a4691ba62032c8cda0fca : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcb95bb412ae458daabe07b4bbfa898f : SUCCESS in 7m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/10e52d86fd8f4a6e8ea261504523eefe : SUCCESS in 12m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b1100226f982448f9f2ef23d5ea08819 : SUCCESS in 7m 39s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8794a273528c4cc89ee8ca60b5674e18 : SUCCESS in 5m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/25209f1c3aae49c691e27bd56d35a314 : SUCCESS in 1h 03m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e396772e49c44013a0b79c628588b8cb : FAILURE in 1h 08m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/734423d955f04e3395b2975885b18a9c : SUCCESS in 1h 20m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/660354d24ed24480be22b155b6292ec8 : SUCCESS in 1h 24m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/695d39899b0246a29893b076c23b609e : SUCCESS in 1h 08m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fa43c9a7b1af47109a105b6d9607843d : SUCCESS in 1h 15m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2fda257adfa2d4bf0c75f387d97acf4b7d1c3fca","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-12 05:00:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ca772bedc494e066b4ac0ec4a27aa8d426f8cd00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 06:15:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/72dd982f04a34cf1802129494b1d8fac : SUCCESS in 45m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b7773d29608441a0a7e3163c95cd66d7 : SUCCESS in 50m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbe38d9c557d4b37b399f7644b652d60 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5826f65bea14fd79478ee1ed37b9758 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f00e0216409f4b3fa77cb8f4c1aa18fd : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecac3e4ccc11445081290b501f0da3ea : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c7722e08e9043cf861f9ea782c88690 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2da21322c8c4d86a2bb92619e02d2d9 : SUCCESS in 5m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/673439bff3894649b50cbcb4ad634b50 : SUCCESS in 8m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7ad0068d1abe48668bd9a35346137a24 : SUCCESS in 9m 01s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5f166ca7dcd04887886c32781ab2b78f : SUCCESS in 3m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3685ed3705fe4a6ba05efd40e11f7267 : SUCCESS in 1h 11m 27s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fd9edc4b40be47beae751828f78e4c9b : SUCCESS in 1h 03m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/482b2e9ed434441bb91bb5239372d67f : SUCCESS in 1h 13m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1222f7483d13461cb3c285e816427703 : SUCCESS in 1h 02m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f18b5408fc8a43198086515c958d6bed : SUCCESS in 45m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/12f1f95841444b61bec3d18dbc51e5a2 : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":1},{"id":"6142226e413d967235ad5d0533c34f2cd15e098d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-15 19:24:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"907fc15157d1d88398a15d641d92930821e95f89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 03:24:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/568ed8779fe94ec7a65c275aae48ef53 : SUCCESS in 1h 04m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/613cfdaeded64daa8da92d4ac65d2117 : SUCCESS in 1h 06m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0af3fff2ac94ee896646d4a1cd195cf : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b306c875669a4961a13447a5f14f2aac : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff9090bdbe774e26b705dc42f1db2450 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c71722ae89bf4d1da173a55f3c43963b : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f043f07ac2e4becb5c40651819ead79 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d03309e73d014270a3a4b3756b09c939 : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ffa96c3593724efa9b2976ba490b1bab : SUCCESS in 8m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a6ac558bab7a4da188844c3c8959a45d : SUCCESS in 8m 32s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ec09e614abc94c2cbf105f884b1eb5d9 : SUCCESS in 4m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c248b787cef5412abb4aa88c0e9e44fd : SUCCESS in 1h 02m 40s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d7793ac2377546e1a0fd5e8cfbe3b9cc : SUCCESS in 1h 12m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/711cc086d13e4a6a86c9f17c030d2f7f : SUCCESS in 1h 11m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dafd1336a50e4b48a271d99b15052944 : SUCCESS in 1h 12m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e0bd0d7e44842229ced8bd3f6bc1cb5 : SUCCESS in 52m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/084e84ebbc5c4b2d90355dc1393f7379 : SUCCESS in 1h 24m 48s","accounts_in_message":[],"_revision_number":2},{"id":"b18e06128ca1b14036f0b2b3b52eacb2c4ad8e4a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-16 07:07:50.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"1d1b2aabc7a96003112485ac8eba937e9f4028f7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-16 15:45:37.000000000","message":"Topic bug/1915543 removed","accounts_in_message":[],"_revision_number":2},{"id":"fb82e5070f09eadca5e6671fffe7f936eec58924","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-16 15:45:50.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":2},{"id":"7d354c995c5b4fcda17ec4b170bcd79ef7888b81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-19 08:34:51.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"20a13f1dd834d4fb9d1af0d4b6eb7c290af31925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 10:51:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d44a1bf892ac4849a64795350ab15747 : SUCCESS in 57m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/149180188e184fc58db6fbfa99819eb7 : SUCCESS in 1h 16m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/382938ad4e534bf79c19cdff20e0373a : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcec7655be9e42b8a7f8e5b01d795037 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e07ccff80994e4eaaddfaa2b9948290 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d492edc2bf564e078ac3a53201cffb90 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aed98de0694e40ec91d8f99f3323ffcb : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ced186a87f84530b2da346e943dec67 : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d6c3876e5a6347c9b394dede7e06e01a : SUCCESS in 11m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c202fd65d3104908ab88d86dd121fe25 : SUCCESS in 12m 30s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/dc30c9bd0a6648c29518dc532edfac80 : SUCCESS in 5m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/45ee6298a4884e3c99e2882982e082fc : SUCCESS in 1h 00m 11s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/794f4f8dc40a47998ab499b243286551 : SUCCESS in 1h 14m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/37de8c3a0bf14328898ad84107a86c2b : SUCCESS in 1h 11m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/512b61aa9ef247f5bd4ddc9ca3e9220d : SUCCESS in 1h 09m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d547dcec33fc43f7800c649761dbb0b2 : SUCCESS in 1h 02m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/74f16e4fcf074c969206d18db3dbd1c2 : SUCCESS in 1h 09m 11s","accounts_in_message":[],"_revision_number":3},{"id":"4109f42e73cbb2d3a030678152552eef635382b2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-19 14:47:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"63d2ee1893da5152bc9d4e7d064ea8e0706b0347","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-19 17:55:40.000000000","message":"Patch Set 3:\n\nWe\u0027re in a race here with https://review.opendev.org/c/openstack/devstack/+/774524\n\nBut if that ends up landing first we obviously don\u0027t need this and I\u0027ll clean it up.\n\nI just needed to have this in the chain somewhere to get things working in tempest with devstack.","accounts_in_message":[],"_revision_number":3},{"id":"343917c9a8a0dbbe69255ef2e96358287e735315","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-22 18:03:54.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"b1576667e72221566f68202ce73e0c31316db9a3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-22 18:05:50.000000000","message":"Topic secure-rbac-handle-members removed","accounts_in_message":[],"_revision_number":4},{"id":"226a3653b0990ef2b4cc808d1664315c1e68713c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-22 18:05:57.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":4},{"id":"7817d9b734c67469e2df1f50903bc63ee3cf72e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 21:30:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b139c56fdede421fa939b9c93460ff64 : SUCCESS in 1h 04m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d0f38d9db2fe459d9fb885cb54de523e : SUCCESS in 1h 39m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8c3d69e4f9d4d04850cdb5c95e633ff : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9800b56978314caf9eb9acbb2ac81d13 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/600ab89cf7bc42b889c8455d1e6e5011 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37f13cd8479346169568149540ed5402 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19044db398aa424d9664f4d0cf6a850e : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beeb81e8892349b0b30c5f8e5c7c43fc : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2650e28dc616454290fec71cd354fd13 : SUCCESS in 8m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/72f7892873174c648f40e583922b4fc4 : SUCCESS in 14m 08s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/68d829c2cf554119917e005b11f4b235 : SUCCESS in 6m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a1e0c5c468284b5fb32604eab577002f : SUCCESS in 1h 51m 32s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7164d1dfebe2448daa989c446d284ac0 : SUCCESS in 1h 21m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2b36a37fb8d74b1d9e25c00f39dced8a : SUCCESS in 1h 19m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/860346811d3544b59e824e07a33f0d04 : FAILURE in 1h 15m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab877312182944548ab357808ab0f6bc : SUCCESS in 1h 08m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7d2d7bcd5cfe44a3a8b2a7c21f578647 : SUCCESS in 1h 54m 29s","accounts_in_message":[],"_revision_number":4},{"id":"eaa6eec27f2bd6d86fad94b0923b7c3ca2968532","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-23 14:50:20.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"672e1cf3a1abc7c32cd19cc6e694962fbe7303be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-23 14:51:00.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"10ece4726bdce36728b335c8ed5d4b4144d0aaa1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-23 15:28:56.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/774308/774308/6/check-not-vote/ext-glance-zuul/b6b8767 : SUCCESS in 37m 06s","accounts_in_message":[],"_revision_number":6},{"id":"6296d02e36bbe7f13cb3b59d085a19937f28f191","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-23 17:31:15.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent, and commit message as Patch Set 6.","accounts_in_message":[],"_revision_number":7},{"id":"2e114ef2f4248c8190c3a8a0f63fc29496ffe6ed","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 18:18:16.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nAbhi was +2 before the several rebases and this is quite trivial, so I\u0027m going to just fast approve so we can start cutting this series down a bit.","accounts_in_message":[],"_revision_number":7},{"id":"0b2979036ed5a6e015c414a7a1ca1e454c85b7fc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-23 22:23:53.000000000","message":"Abandoned\n\nThis isn\u0027t needed now that https://review.opendev.org/c/openstack/devstack/+/774524 landed - and it\u0027s just extra resource churn for nothing.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"e1cd1f49021b8aa81dc67fc6d691594daf158d20","revisions":{"e17cf2a20eda7edc2c729380984a64df3611feda":{"kind":"REWORK","_number":1,"created":"2021-02-05 21:15:14.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/08/774308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/774308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/774308/1"}}},"commit":{"parents":[{"commit":"089eebc1b33f432efffae4ec95a64a1e3644f9cd","subject":"Bump requirements to perpare for secure RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/089eebc1b33f432efffae4ec95a64a1e3644f9cd"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-05 21:12:24.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-05 21:12:24.000000000","tz":0},"subject":"Explicitly unset tempest_roles in devstack plugin","message":"Explicitly unset tempest_roles in devstack plugin\n\nWe need to do this in order to use dynamic credentials in tempest\nwithout having the \u0027member\u0027 role added to personas we only want to have\nthe \u0027reader\u0027 role.\n\nChange-Id: I90d97c7181bd2d449615732a61bd3d8074e93adc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e17cf2a20eda7edc2c729380984a64df3611feda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e17cf2a20eda7edc2c729380984a64df3611feda"}]},"branch":"refs/heads/master"},"f3956cc2bf8be5d120db2287555246b5e0d0ea52":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-15 19:24:04.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/08/774308/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/774308/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/774308/2"}}},"commit":{"parents":[{"commit":"153cbf95471ffe9e6e59245c8a7bb71f41faf789","subject":"Bump requirements to perpare for secure RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/153cbf95471ffe9e6e59245c8a7bb71f41faf789"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-05 21:12:24.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-15 16:26:02.000000000","tz":0},"subject":"Explicitly unset tempest_roles in devstack plugin","message":"Explicitly unset tempest_roles in devstack plugin\n\nWe need to do this in order to use dynamic credentials in tempest\nwithout having the \u0027member\u0027 role added to personas we only want to have\nthe \u0027reader\u0027 role.\n\nChange-Id: I90d97c7181bd2d449615732a61bd3d8074e93adc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3956cc2bf8be5d120db2287555246b5e0d0ea52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3956cc2bf8be5d120db2287555246b5e0d0ea52"}]},"branch":"refs/heads/master"},"3824f4d3009aebb3ad51d70ed49a1e177e885fe6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-19 08:34:51.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/08/774308/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/774308/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/774308/3"}}},"commit":{"parents":[{"commit":"80fa4cfb6e13dd00c338d14d1827b41eb22815a4","subject":"Bump requirements to perpare for secure RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/80fa4cfb6e13dd00c338d14d1827b41eb22815a4"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-05 21:12:24.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-19 05:50:49.000000000","tz":0},"subject":"Explicitly unset tempest_roles in devstack plugin","message":"Explicitly unset tempest_roles in devstack plugin\n\nWe need to do this in order to use dynamic credentials in tempest\nwithout having the \u0027member\u0027 role added to personas we only want to have\nthe \u0027reader\u0027 role.\n\nChange-Id: I90d97c7181bd2d449615732a61bd3d8074e93adc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3824f4d3009aebb3ad51d70ed49a1e177e885fe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3824f4d3009aebb3ad51d70ed49a1e177e885fe6"}]},"branch":"refs/heads/master"},"66b40788150b6225df8618c48d42bc69e7d2aa29":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-22 18:03:54.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/08/774308/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/774308/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/774308/4"}}},"commit":{"parents":[{"commit":"97613e24b90e7f7586d4e232a99cf170812ca3af","subject":"Bump requirements to prepare for secure RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/97613e24b90e7f7586d4e232a99cf170812ca3af"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-05 21:12:24.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-22 17:37:09.000000000","tz":0},"subject":"Explicitly unset tempest_roles in devstack plugin","message":"Explicitly unset tempest_roles in devstack plugin\n\nWe need to do this in order to use dynamic credentials in tempest\nwithout having the \u0027member\u0027 role added to personas we only want to have\nthe \u0027reader\u0027 role.\n\nChange-Id: I90d97c7181bd2d449615732a61bd3d8074e93adc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66b40788150b6225df8618c48d42bc69e7d2aa29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66b40788150b6225df8618c48d42bc69e7d2aa29"}]},"branch":"refs/heads/master"},"11ee076288fbb0d4f60ea867025605be55e4556e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-02-23 14:50:20.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/08/774308/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/774308/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/774308/5"}}},"commit":{"parents":[{"commit":"97613e24b90e7f7586d4e232a99cf170812ca3af","subject":"Bump requirements to prepare for secure RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/97613e24b90e7f7586d4e232a99cf170812ca3af"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-05 21:12:24.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@redhat.com","date":"2021-02-23 14:50:03.000000000","tz":0},"subject":"Explicitly unset tempest_roles in devstack plugin","message":"Explicitly unset tempest_roles in devstack plugin\n\nWe need to do this in order to use dynamic credentials in tempest\nwithout having the \u0027member\u0027 role added to personas we only want to have\nthe \u0027reader\u0027 role.\n\nImplements: blueprint secure-rbac\n\nChange-Id: I90d97c7181bd2d449615732a61bd3d8074e93adc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11ee076288fbb0d4f60ea867025605be55e4556e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11ee076288fbb0d4f60ea867025605be55e4556e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c477c499ba5f67efb1b693cb814d9abf4d632db4":{"kind":"NO_CHANGE","_number":6,"created":"2021-02-23 14:51:00.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/08/774308/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/774308/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/774308/6"}}},"commit":{"parents":[{"commit":"a55660c84afbc4105d0e098f7d386df06da72d4b","subject":"Bump requirements to prepare for secure RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a55660c84afbc4105d0e098f7d386df06da72d4b"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-05 21:12:24.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@redhat.com","date":"2021-02-23 14:51:00.000000000","tz":0},"subject":"Explicitly unset tempest_roles in devstack plugin","message":"Explicitly unset tempest_roles in devstack plugin\n\nWe need to do this in order to use dynamic credentials in tempest\nwithout having the \u0027member\u0027 role added to personas we only want to have\nthe \u0027reader\u0027 role.\n\nImplements: blueprint secure-rbac\n\nChange-Id: I90d97c7181bd2d449615732a61bd3d8074e93adc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c477c499ba5f67efb1b693cb814d9abf4d632db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c477c499ba5f67efb1b693cb814d9abf4d632db4"}]},"branch":"refs/heads/master","description":"Rebase"},"e1cd1f49021b8aa81dc67fc6d691594daf158d20":{"kind":"NO_CHANGE","_number":7,"created":"2021-02-23 17:31:15.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/08/774308/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/774308/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/774308/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/774308/7"}}},"commit":{"parents":[{"commit":"764f29a99d27edb2d8bf9d299deabcdb84ac08d6","subject":"Bump requirements to prepare for secure RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/764f29a99d27edb2d8bf9d299deabcdb84ac08d6"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-05 21:12:24.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-23 17:27:09.000000000","tz":0},"subject":"Explicitly unset tempest_roles in devstack plugin","message":"Explicitly unset tempest_roles in devstack plugin\n\nWe need to do this in order to use dynamic credentials in tempest\nwithout having the \u0027member\u0027 role added to personas we only want to have\nthe \u0027reader\u0027 role.\n\nImplements: blueprint secure-rbac\n\nChange-Id: I90d97c7181bd2d449615732a61bd3d8074e93adc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1cd1f49021b8aa81dc67fc6d691594daf158d20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1cd1f49021b8aa81dc67fc6d691594daf158d20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
