)]}'
{"id":"openstack%2Fglance~774314","triplet_id":"openstack%2Fglance~master~Iebbb2dcb767ecf3c965f34f1ca04af20a2039be1","project":"openstack/glance","branch":"master","topic":"bug/1914826","hashtags":[],"change_id":"Iebbb2dcb767ecf3c965f34f1ca04af20a2039be1","subject":"Make web-download revert all stores on fail","status":"MERGED","created":"2021-02-05 22:24:41.000000000","updated":"2021-03-06 17:33:58.000000000","submitted":"2021-03-06 17:33:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"774314-bug/1914826","meta_rev_id":"559099d4632cc836c18377525a1dcd5eb648801e","_number":774314,"virtual_id_number":774314,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-06 17:33:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-02-09 13:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-03-05 18:23:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-02-10 05:34:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-05 23:50:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-08 16:48:19.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-09 13:56:08.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"9440334f994adb8bdd540b1e3d80620c6231a3cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-05 22:24:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c78539dbe3facaa687a6571f1ff1765815d8cd22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 23:50:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5d8e72d97bcd4ddcb1ef97a5700d07de : SUCCESS in 1h 04m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0d26b06d90f74d058b82707c0127f0a3 : SUCCESS in 59m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1133e64630be4578867e1fea6359ea8e : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/015d43fd79c4422ab3ddf1562e9d97d3 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de14f53231354e1591430f1cbbb59dd2 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f83cfd19009649c091810329ef0f5c2a : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8be79e27cb2d409ca70f76c6481c3344 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45aef53f7ff9478092f79003984424a1 : SUCCESS in 7m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d2217f8f31284ac9ae17389d179581b8 : SUCCESS in 7m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4444ba43695d4388abd9297b7324de94 : SUCCESS in 7m 39s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c3573f1bff5f477b83284b7e7fbaa3d1 : SUCCESS in 5m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6017526522f84eec94fbaf0b80ea5e33 : SUCCESS in 1h 13m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c2c159cec7264cb492c12fe82df28ecd : SUCCESS in 1h 22m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/90722e30bdfc41b183d7e17c34ab6c39 : SUCCESS in 1h 20m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/19fe2cd14f5f459c9e60c22fc036e9c9 : SUCCESS in 1h 17m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/313ed1c87a8d49fd8d6fa3959ef81884 : SUCCESS in 1h 06m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b8a856c252f54b198e955b125f9fd057 : SUCCESS in 1h 03m 01s","accounts_in_message":[],"_revision_number":1},{"id":"4d5ce0d95adf13e72c9e5236a65cf0289d792e1c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-08 16:41:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a1a3409f000674edc5498e6a481fa802d5211ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-08 16:46:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"56d76794b51fc19b523cf98b5a1e5857c69c6ef6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-08 16:48:19.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"fed4b50e08ea8c2457dc6ff36eebe07125e44d1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 19:21:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/dbae1def17244353bc65aba3926e8ae4 : SUCCESS in 1h 03m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5f1a05957d884fb2b3ec5a0805cfb063 : SUCCESS in 1h 12m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97d992865fa0404c823cdc7eec54bb2d : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5a80a8491e64603a1de5b118c7b14dc : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b290a09594c64fa58c2a56812f369dd8 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85c7e105792c4bd6b0a64b7b3817255a : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/480a2e319b7044f395ddecb59acba265 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21810117754f4b30877b594d68669791 : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bdaa0881d5b34750a42a225ebf06c23d : SUCCESS in 8m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fbe209922b5a4001bde711c3588422bc : SUCCESS in 7m 43s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5326af8e13494a50b7ac52d838da361b : SUCCESS in 4m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2625a6ae7bed4da8b3ccf5cc90c87866 : SUCCESS in 1h 01m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/be8a77e197e842abb706ca9e51cf41fd : SUCCESS in 1h 29m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ac26b5e51a2041409bf0e3ec7531fc7b : SUCCESS in 1h 20m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/84dafb8f414747c2abcddf35425959a6 : SUCCESS in 1h 13m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f881e051a78e428580066c980fe9b8ab : SUCCESS in 1h 05m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4294cc14b92d4244b305a8185258f9fd : SUCCESS in 1h 15m 51s","accounts_in_message":[],"_revision_number":2},{"id":"b9fe404c593f364cfc1cee35572ac8b7cff8022f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-09 13:56:08.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nThank you dan, looks good to me","accounts_in_message":[],"_revision_number":2},{"id":"313517e4acb8bf14db3f831c0f1a5ab7b5e1f793","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-09 15:37:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63d61f86913f7768348e35b161d5c4989fc08a4d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-10 05:34:44.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"59bc4bdd40c900b65ab674737bd40a0c4ca51800","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-05 18:06:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae9dff511639c52a7b8c28f241e93b1eb8fda2ef","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-05 18:23:40.000000000","message":"Patch Set 2: Workflow+1\n\nNinja approving, sitting open for long, straight forward fix.\nThank you.","accounts_in_message":[],"_revision_number":2},{"id":"1a2da7e9c95f1d5a38ff8f1346ad4cda83648405","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 01:55:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"446196f5e5ad5df665db20d535174d482d7da02d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 11:41:38.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"9f89ba2085f3616cac01d5e6cfc9384850c8d6a0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-06 11:47:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"105fe756c488ee1968f42b6bbfac619e4d27b8e2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-06 12:23:25.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/774314/774314/2/check-not-vote/ext-glance-zuul/7cda203 : SUCCESS in 35m 38s","accounts_in_message":[],"_revision_number":2},{"id":"926aafc7d7c87da9a457cf5528cce494aa1c646b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 13:09:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/53782a25644c4640b69ffec3b810fa6c : SUCCESS in 1h 02m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/049120324d734df8953831e690ba491c : SUCCESS in 1h 14m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8ff634a3b7440fe93b5e4eec7d614d1 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2137850f1c964759bd1f1420c2c10919 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f7f4d94d09e471080712ff90550a669 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a22ba43e6474423b33c2146d75a3704 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11e337f265e248f7b1a3436ed06a5d4e : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61a6ae08bd6c4436b35e291aea6e2bff : SUCCESS in 6m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/87a557c7d2e34add9691eb0665716ffa : SUCCESS in 9m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4fa0c21fd9af4defb413f8bd8b21660a : SUCCESS in 13m 45s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/641ddc3c43d64009b833e1a54c031e1d : SUCCESS in 5m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7ac5167a0e674b3cbddf5b6b42509ecf : SUCCESS in 1h 12m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7cb49de4918845bcabe274df531256eb : SUCCESS in 1h 07m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8397979d76ee4f26a0b0b5ef6dc78876 : SUCCESS in 1h 07m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d48521bfcd634419953eccf161870ebb : SUCCESS in 1h 07m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eac34b18efe54cb8ae65f65f8f679ff6 : SUCCESS in 58m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9ca8191df66e43cf83c73ae94fd5413d : SUCCESS in 1h 06m 33s","accounts_in_message":[],"_revision_number":2},{"id":"096bf5ce099e78f1ab8336990a4b69b0a702a8f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 13:22:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3dd6543e087faf3ed8b700ad9341ed68451cfc49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 17:33:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bdaf28abc16f4496a797695361b4c9f8 : SUCCESS in 1h 03m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3593e34b9494c1a8e5f998cbcf15be1 : SUCCESS in 1h 14m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cea26c94b0a44a0a9a652c01bcd16d5 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f04fe0d14ae4652b99f62bd8fc9dda1 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/834fe67840da4280912c87cbbaed88f0 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4ce7736ba134deab63f957a37ddec29 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd00bfbcbcfc4a8d8196115c4bc01c80 : SUCCESS in 8m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f82a8e677dd54f0fb5c30b09ea1870be : SUCCESS in 8m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1396b02eca5e45f6950b900f40c80e43 : SUCCESS in 9m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bd431d625b5047a29730e60c367b04b8 : SUCCESS in 1h 16m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e58d07a44d6c4577b87f4bd77853d69b : SUCCESS in 1h 20m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bde0434761449cbbc000bb0577c3b52 : SUCCESS in 55m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cc002ae56c2c44dfbfccc1a164b105bf : SUCCESS in 1h 13m 00s","accounts_in_message":[],"_revision_number":2},{"id":"559099d4632cc836c18377525a1dcd5eb648801e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 17:33:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cbb2af6e329c1bf6c1c39cb59f95aaad2c0d7a07","revisions":{"82f529e31a4df58163b390216228aef5c996353b":{"kind":"REWORK","_number":1,"created":"2021-02-05 22:24:41.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/774314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/774314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/774314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/774314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/774314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/774314/1"}}},"commit":{"parents":[{"commit":"68646db35d4a389237b9c1db75c074ba90fa1ec6","subject":"Pass ImageActionWrapper to internal plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/68646db35d4a389237b9c1db75c074ba90fa1ec6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 22:35:51.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 23:18:21.000000000","tz":-480},"subject":"Make web-download revert all stores on fail","message":"Make web-download revert all stores on fail\n\nIf web-download fails to stage the image, no subsequent import to\nstore tasks could have run. Thus, we should remove all of them from\n\"importing\" and add them to \"failed\" so that it is obvious from the\noutside that we failed.\n\nThis is another good reason for having the $image/tasks API, but we\nalso should not continue advertising \"yes we\u0027re importing to $store\"\nwhen we are not.\n\nChange-Id: Iebbb2dcb767ecf3c965f34f1ca04af20a2039be1\nCloses-Bug: #1914826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/82f529e31a4df58163b390216228aef5c996353b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/82f529e31a4df58163b390216228aef5c996353b"}]},"branch":"refs/heads/master"},"cbb2af6e329c1bf6c1c39cb59f95aaad2c0d7a07":{"kind":"REWORK","_number":2,"created":"2021-02-08 16:46:28.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/774314/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/774314/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/774314/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/774314/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/774314/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/774314/2"}}},"commit":{"parents":[{"commit":"68646db35d4a389237b9c1db75c074ba90fa1ec6","subject":"Pass ImageActionWrapper to internal plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/68646db35d4a389237b9c1db75c074ba90fa1ec6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 22:35:51.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-08 17:43:14.000000000","tz":-480},"subject":"Make web-download revert all stores on fail","message":"Make web-download revert all stores on fail\n\nIf web-download fails to stage the image, no subsequent import to\nstore tasks could have run. Thus, we should remove all of them from\n\"importing\" and add them to \"failed\" so that it is obvious from the\noutside that we failed.\n\nThis is another good reason for having the $image/tasks API, but we\nalso should not continue advertising \"yes we\u0027re importing to $store\"\nwhen we are not.\n\nChange-Id: Iebbb2dcb767ecf3c965f34f1ca04af20a2039be1\nCloses-Bug: #1914826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cbb2af6e329c1bf6c1c39cb59f95aaad2c0d7a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cbb2af6e329c1bf6c1c39cb59f95aaad2c0d7a07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
